)]}'
{"id":"openstack%2Fmetalsmith~896684","triplet_id":"openstack%2Fmetalsmith~stable%2Fzed~Id258b294972c2f5ea816cb274773b8947f92e110","project":"openstack/metalsmith","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-09-28 13:14:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id258b294972c2f5ea816cb274773b8947f92e110","subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","status":"MERGED","created":"2023-09-27 15:59:01.000000000","updated":"2023-09-28 13:15:49.000000000","submitted":"2023-09-28 13:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896684","meta_rev_id":"fe855ca67b6b3c43855fa028469af9b7c1a2ee44","_number":896684,"virtual_id_number":896684,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 13:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2023-09-28 10:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-28 11:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-28 11:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 16:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 10:18:44.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2023-09-28 11:32:48.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"213dcd534d82ef5b8759e48404ab1aea82b6862e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 15:59:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31e58cfd004bbe781b8e80e2bae0d7d642e47d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:44:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd727eb904444956b3b90dbc34ce2abc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1449bad6664eff98194cd21959d335 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/502151671a9648289e87b91533f4350e : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95fdf838fc634ec0ac0006d925feaf28 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bd2bad3427e4f7286422e1517c2744b : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af6e5286b5004b9e91ec5ec0f232896f : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2a4ad197504f608482adefbc7462b9 : SUCCESS in 5m 53s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/9c80efa0af42475588384eba745e285c : FAILURE in 23m 56s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/2f185dd9f60e4d1089f841879d6ef751 : FAILURE in 40m 19s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/837c691ef45c4c48a63a923de87d7f04 : FAILURE in 39m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c2de4b737ad4bb1b4007636eb5788e2 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"47d165c8c3ad26c923b18c13fa51a7ae83834e33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 16:59:50.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4c2c772a3557efb71633c3e565ce9b8bea91bb1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 17:46:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/166b3b2f77f24a3a8113bd8a104fb301\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc353a7deb2e4d3b87115b22d7e6bfc8 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1e61fabcd2a4eb0818a159fc69e7433 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/558c78f11bef4dce8f8c5fba4eb9e327 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef2df0251c5f494d9ea7ff36809ba713 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc1fccda724f4b9d8c08346eaa0f8f3c : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7188a05ddacf491f800dbb3c390bac4d : SUCCESS in 7m 12s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/e416abdeccb74b72a29340ec01591d70 : FAILURE in 41m 19s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/cec0c6c18d18418f9a851689bc8e2686 : FAILURE in 27m 58s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/c4b55c7f76eb4166b181fed39b10ca06 : FAILURE in 25m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce9669434bc04abb887e6fcc432e3725 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0e5596d28c106f33a49801e8d8b0c73d861d9ed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 18:04:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e49655cab8ee4adc94026ebb339c9c7b066a0cb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 18:17:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"92482e70ba9016241143321915b70bf7e93d2c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 19:14:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3426329b05d54866a5690d247ac00932\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b0c9659c29443d86eb30471d9f51ad : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309e13910310451c9e754fd093bddbd3 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebacf5197a154d7186de29bcb5573fa4 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7405a0dd0ebf469aa468b2ef510cee0a : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/401a59ecdc454a8eb9bf2842011bb0a2 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b873a1f7afa34ddda75e1e04468e04c3 : SUCCESS in 8m 01s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/dc68b35352374ba09e5be5e8d0de0596 : FAILURE in 28m 52s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/eacf1f0867f54176af5f81c93cc7364c : FAILURE in 50m 34s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/98eda0679ab84ac0b3acf0da68d217ff : FAILURE in 28m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/069b3659e8a74f1da752f5e9ac40153f : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":4},{"id":"928c9b0d120d4e72aded3d23b7ebbe6e269818fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 20:47:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"954f6231dcf4e4e758186daf88be0b003ec84dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 21:30:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3e4315cfd484a31857a406fc82dd88a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a98ae52be0443dbda766fe8dda0f5b : SUCCESS in 2m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52d1d8426db5450ba35a4210cc9881f1 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81becb1d6ced43f7b1cc202161f922be : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9391657d2bc14347ac08dcca52da7dec : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00eef26eed684136b0ebcdda78a6a9b4 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83464a6efe51440cb2c00e9b597f70f8 : SUCCESS in 5m 52s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/c0b014b7aabf4663b16f6c88dc0c1ae4 : FAILURE in 29m 13s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/303d2e1cd40f48e6a51b77ab46f0c637 : FAILURE in 37m 50s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/cbc947ed431e457c80daf2701c52acdc : FAILURE in 25m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1926f3b4258347a78d13c8f3ad96c37d : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":5},{"id":"e1131f79e90433c990ba1cb1efa6e2c60ecc0455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 21:59:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"efd8c0830597da63e1910e7a63bdc66894c2bb7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 22:53:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca0e8b1e325b480f8b9de0a88223aab3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89704e2f677341be93ecd809b82d3a0b : SUCCESS in 2m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05e2c40d8f224e11b91e3505a3b53109 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf3cbedd9b944f54ae285e3fa2df2824 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c681d3a8fa2d4a1f8ec6fe3262a2d816 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee839e03bb834aff8ce65d1acc878106 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc3cbfe933b47dd9d17bc1fbea5a3fd : SUCCESS in 6m 02s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/1a9e438db59d4e76bb5a9c771837ab11 : FAILURE in 47m 28s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/3d27d23d88304165a7f25c603bb8e6e1 : FAILURE in 48m 25s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/4424d263e5664c5ea4c06d08d587f349 : FAILURE in 40m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d7902abe423141e3bfd4e67903b660de : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":6},{"id":"80f3b592a230a556011d06827d386f78f5db0207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 22:55:17.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"bbcaf4d50e0910dc7f029cf64db897c145ff7cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 00:28:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29a054bdb4024712b2ffc90973cf2dee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da5faf6f2a243dc8c057adbc3c7c335 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deb8da18916643fcbb21d57afb718beb : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b7ea54769cb43a6bb5c405ca64800ba : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4457ca394a374fbe8b8afccd85f6b5af : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37f9fb553a134237a7a99b8402a0ebb3 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0905198adb534b73bf799dd0453f345e : SUCCESS in 6m 51s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/c8fa81184fc84b69b688a9820232d3e3 : FAILURE in 1h 26m 03s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/a5f4759f33ec4f9784ddf9495b3577c1 : SUCCESS in 1h 28m 14s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/0ef9bb67a83a40749443ffb10bd0af2f : SUCCESS in 1h 16m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8eeafbea7e4347edbe84cb71e69816e8 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":7},{"id":"d5d5b2ee077c2b09e12dd87bb9cc739c080b4b7d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-28 00:39:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33206f58ca32dafa57374766dbfffc762503c829","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-28 02:29:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4c44ef6fd6908a9424dfc4ce424e711221435928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 04:21:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c150416d58d4cc086676d9448906427\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f236288e93ea4612807aecbb97b5c0a4 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b692643073841b3ba934fbaf28bd2a3 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d28269da57647388739721c9d4207c8 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d612f1181eef4f94974b8e74d5c3597e : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b91827199dd44a4f8628750ecacc91e7 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09aa0c5e917f440c87b767abf9299937 : SUCCESS in 6m 12s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/0888da75b90f4650870befaabd63f1f5 : SUCCESS in 1h 08m 45s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/40e12804f23f4085af49b9348b6bbbfe : SUCCESS in 1h 32m 50s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/f4eb6ce561ad4f4bb43e8b84d2017730 : SUCCESS in 47m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/551787681d8f4ed2b6640fa3f32145ab : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":7},{"id":"6caab30241f329c38a2640e8f73723431cfa912c","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-09-28 10:18:44.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"78794961e71d0cc51db731f9cfe0f74cc3f12975","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-09-28 11:32:48.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"741a48795afa7eb029b00fc5ecf523a51520797d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:33:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ee12c172cf89aded1a9a96a95781c5ea3c141682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:14:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98a3eb8f5a854bb4a6ece083423635fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae5557397684616aea3c601f85f007c : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bea57e2a5d34972881b25a3603f12df : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94a751b2ad434a2795becc1e0f7d15be : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7900ccf8cf0f4e818c0f681eae42d466 : SUCCESS in 6m 34s\n- metalsmith-integration-glance-centos8-uefi https://zuul.opendev.org/t/openstack/build/fc4edb740950417299e220370e54da89 : SUCCESS in 1h 36m 55s\n- metalsmith-integration-glance-centos8-legacy https://zuul.opendev.org/t/openstack/build/f541690063424512bf395ebfc718436a : SUCCESS in 1h 28m 43s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/f6da233d1b744a6f93397f71e06b1b47 : SUCCESS in 1h 14m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4de5139f4cec4f07a70c00c644894a2a : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":7},{"id":"c0b81d912badc58632bf18bf4d46b8e7e8c0f7b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:14:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"fe855ca67b6b3c43855fa028469af9b7c1a2ee44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:15:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33ccd0111b6d4b5f8cf002f184019882\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfa15a68990462d8da7f4ea37f85083 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"47f5953adf3b8912b7503dbba9d6a361ba91d296","revisions":{"63a6d38c898b1432c8afd0b13e923ae62b05170a":{"kind":"REWORK","_number":1,"created":"2023-09-27 15:59:01.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/84/896684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/84/896684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/1"}}},"commit":{"parents":[{"commit":"176f48ed8671656d6d5a20e37215b849ac0978c1","subject":"stable-only: Constrain the upper Ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","message":"Stable-Only: CI: Force install an appropriate openstacsdk version\n\nAnsible has a weird constraint, of \u003c 0.100.0, and doesn\u0027t recognize\nopenstacksdk 0.101.0 or newer for some versions of ansible. While\nothers, don\u0027t recognize versions newer than 1.0.0.\n\nAs such, our only real choice is to force lock the version to\na more version/cycle appropriate version to ensure compatability\nsince we also can\u0027t constain via requirements at this point.\n\nChange-Id: Id258b294972c2f5ea816cb274773b8947f92e110\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/63a6d38c898b1432c8afd0b13e923ae62b05170a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/63a6d38c898b1432c8afd0b13e923ae62b05170a"}]},"branch":"refs/heads/stable/zed"},"19a3f5b8838779774be0e2fd4a10f37121b2f4b3":{"kind":"REWORK","_number":2,"created":"2023-09-27 16:59:50.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/84/896684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/84/896684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/2"}}},"commit":{"parents":[{"commit":"176f48ed8671656d6d5a20e37215b849ac0978c1","subject":"stable-only: Constrain the upper Ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 16:59:50.000000000","tz":0},"subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","message":"Stable-Only: CI: Force install an appropriate openstacsdk version\n\nAnsible has a weird constraint, of \u003c 0.100.0, and doesn\u0027t recognize\nopenstacksdk 0.101.0 or newer for some versions of ansible. While\nothers, don\u0027t recognize versions newer than 1.0.0.\n\nAs such, our only real choice is to force lock the version to\na more version/cycle appropriate version to ensure compatability\nsince we also can\u0027t constain via requirements at this point.\n\nChange-Id: Id258b294972c2f5ea816cb274773b8947f92e110\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/19a3f5b8838779774be0e2fd4a10f37121b2f4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/19a3f5b8838779774be0e2fd4a10f37121b2f4b3"}]},"branch":"refs/heads/stable/zed"},"a8dfaf26c302deb516d777c48db1f7e6ba9fdaee":{"kind":"REWORK","_number":3,"created":"2023-09-27 18:04:23.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/84/896684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/84/896684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/3"}}},"commit":{"parents":[{"commit":"176f48ed8671656d6d5a20e37215b849ac0978c1","subject":"stable-only: Constrain the upper Ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 18:04:10.000000000","tz":-300},"subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","message":"Stable-Only: CI: Force install an appropriate openstacsdk version\n\nAnsible has a weird constraint, of \u003c 0.100.0, and doesn\u0027t recognize\nopenstacksdk 0.101.0 or newer for some versions of ansible. While\nothers, don\u0027t recognize versions newer than 1.0.0.\n\nAs such, our only real choice is to force lock the version to\na more version/cycle appropriate version to ensure compatability\nsince we also can\u0027t constain via requirements at this point.\n\nChange-Id: Id258b294972c2f5ea816cb274773b8947f92e110\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a8dfaf26c302deb516d777c48db1f7e6ba9fdaee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/a8dfaf26c302deb516d777c48db1f7e6ba9fdaee"}]},"branch":"refs/heads/stable/zed"},"7d13b8a4a947a8bc34173f37acf1e1c659fbdc8a":{"kind":"REWORK","_number":4,"created":"2023-09-27 18:17:13.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/84/896684/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/84/896684/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/4"}}},"commit":{"parents":[{"commit":"176f48ed8671656d6d5a20e37215b849ac0978c1","subject":"stable-only: Constrain the upper Ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 18:17:04.000000000","tz":-300},"subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","message":"Stable-Only: CI: Force install an appropriate openstacsdk version\n\nAnsible has a weird constraint, of \u003c 0.100.0, and doesn\u0027t recognize\nopenstacksdk 0.101.0 or newer for some versions of ansible. While\nothers, don\u0027t recognize versions newer than 1.0.0.\n\nAs such, our only real choice is to force lock the version to\na more version/cycle appropriate version to ensure compatability\nsince we also can\u0027t constain via requirements at this point.\n\nChange-Id: Id258b294972c2f5ea816cb274773b8947f92e110\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7d13b8a4a947a8bc34173f37acf1e1c659fbdc8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7d13b8a4a947a8bc34173f37acf1e1c659fbdc8a"}]},"branch":"refs/heads/stable/zed"},"fb6a7e77ddd869a6b1f45d9959ab0834372c4bff":{"kind":"REWORK","_number":5,"created":"2023-09-27 20:47:19.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/84/896684/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/84/896684/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/5"}}},"commit":{"parents":[{"commit":"176f48ed8671656d6d5a20e37215b849ac0978c1","subject":"stable-only: Constrain the upper Ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 20:45:00.000000000","tz":-300},"subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","message":"Stable-Only: CI: Force install an appropriate openstacsdk version\n\nAnsible has a weird constraint, of \u003c 0.100.0, and doesn\u0027t recognize\nopenstacksdk 0.101.0 or newer for some versions of ansible. While\nothers, don\u0027t recognize versions newer than 1.0.0.\n\nAs such, our only real choice is to force lock the version to\na more version/cycle appropriate version to ensure compatability\nsince we also can\u0027t constain via requirements at this point.\n\nAdditionally, as a result of change I7b9c70b03c17c631da8e53f70808fed103c95c06\nand the fact that the base issue broke metalsmith CI, we need to merge\nthat change in with this one.\n\nChange-Id: Id258b294972c2f5ea816cb274773b8947f92e110\nCo-Authored-By: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/fb6a7e77ddd869a6b1f45d9959ab0834372c4bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/fb6a7e77ddd869a6b1f45d9959ab0834372c4bff"}]},"branch":"refs/heads/stable/zed"},"b5d2930855b15a1ee9f0580aa3ab7863f3751ddd":{"kind":"REWORK","_number":6,"created":"2023-09-27 21:59:41.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/84/896684/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/84/896684/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/6"}}},"commit":{"parents":[{"commit":"176f48ed8671656d6d5a20e37215b849ac0978c1","subject":"stable-only: Constrain the upper Ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 21:59:32.000000000","tz":-300},"subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","message":"Stable-Only: CI: Force install an appropriate openstacsdk version\n\nAnsible has a weird constraint, of \u003c 0.100.0, and doesn\u0027t recognize\nopenstacksdk 0.101.0 or newer for some versions of ansible. While\nothers, don\u0027t recognize versions newer than 1.0.0.\n\nAs such, our only real choice is to force lock the version to\na more version/cycle appropriate version to ensure compatability\nsince we also can\u0027t constain via requirements at this point.\n\nAdditionally, as a result of change I7b9c70b03c17c631da8e53f70808fed103c95c06\nand the fact that the base issue broke metalsmith CI, we need to merge\nthat change in with this one.\n\nChange-Id: Id258b294972c2f5ea816cb274773b8947f92e110\nCo-Authored-By: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b5d2930855b15a1ee9f0580aa3ab7863f3751ddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b5d2930855b15a1ee9f0580aa3ab7863f3751ddd"}]},"branch":"refs/heads/stable/zed"},"47f5953adf3b8912b7503dbba9d6a361ba91d296":{"kind":"REWORK","_number":7,"created":"2023-09-27 22:55:17.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/84/896684/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/84/896684/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/84/896684/7"}}},"commit":{"parents":[{"commit":"176f48ed8671656d6d5a20e37215b849ac0978c1","subject":"stable-only: Constrain the upper Ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/176f48ed8671656d6d5a20e37215b849ac0978c1"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 15:56:37.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-27 22:55:09.000000000","tz":-300},"subject":"Stable-Only: CI: Force install an appropriate openstacsdk version","message":"Stable-Only: CI: Force install an appropriate openstacsdk version\n\nAnsible has a weird constraint, of \u003c 0.100.0, and doesn\u0027t recognize\nopenstacksdk 0.101.0 or newer for some versions of ansible. While\nothers, don\u0027t recognize versions newer than 1.0.0.\n\nAs such, our only real choice is to force lock the version to\na more version/cycle appropriate version to ensure compatability\nsince we also can\u0027t constain via requirements at this point.\n\nAdditionally, as a result of change I7b9c70b03c17c631da8e53f70808fed103c95c06\nand the fact that the base issue broke metalsmith CI, we need to merge\nthat change in with this one.\n\nChange-Id: Id258b294972c2f5ea816cb274773b8947f92e110\nCo-Authored-By: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/47f5953adf3b8912b7503dbba9d6a361ba91d296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/47f5953adf3b8912b7503dbba9d6a361ba91d296"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
