)]}'
{"id":"openstack%2Fmetalsmith~906787","triplet_id":"openstack%2Fmetalsmith~master~Ic7afcf11342e9a33d7fb76128a0c9746ab879dca","project":"openstack/metalsmith","branch":"master","topic":"ironic-codespell","attention_set":{},"removed_from_attention_set":{"36102":{"account":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"last_update":"2024-01-30 22:04:39.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-01-30 20:23:58.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}},"hashtags":[],"change_id":"Ic7afcf11342e9a33d7fb76128a0c9746ab879dca","subject":"[codespell] Adding CI target for Tox Codespell","status":"MERGED","created":"2024-01-26 04:26:43.000000000","updated":"2024-01-30 22:06:28.000000000","submitted":"2024-01-30 22:04:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"906787-ironic-codespell","meta_rev_id":"95f0926f4776ea1086017cec5764d15b9f1ed0d8","_number":906787,"virtual_id_number":906787,"owner":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-30 22:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-01-30 20:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2024-01-30 20:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-01-30 20:23:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 04:27:14.000000000","updated_by":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-01-26 06:10:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-30 20:12:56.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"f1b23f6c2c3e797c16e7f79ec1b448a40ffea7d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-26 04:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7de45f3b93d966eadbb385cf63fb995d49cd409","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-26 04:27:18.000000000","message":"Topic set to ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"561fd72b47b36c765c865de86f7b423eb2165bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 06:10:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e37cd5e4274bdda0e0a1b49c7270f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d65ee9578aa84c0ca4a39d222f5a8f3e : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d928ccd4b9804f98b7ece71e24f8c6d9 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58ea84e5a2ba4eafa782e972dfc01fb2 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7401657e7ad84b60a5a741cbbd10b471 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668b9e15d9ea49c59c1a2d6cf948f622 : SUCCESS in 7m 22s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/2959bc70bda743d5948111e28ea62dae : SUCCESS in 1h 03m 23s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/8ef72837991e4628b9bb8b442bc80267 : SUCCESS in 1h 33m 57s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/ef4e0606235f4c978adf33417fa9dfe0 : SUCCESS in 1h 11m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/734d750c0b1e462ea3961688a7ae3c79 : SUCCESS in 3m 58s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/1a1a6c7505b4410080a2014ad83a5a3d : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"368fc051fb7d9517fa3b0e15ba9063b1c98e2fc3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-27 06:29:46.000000000","message":"Topic ironic-week-prio removed","accounts_in_message":[],"_revision_number":1},{"id":"13fc5929fc5ab2ef03000b49af4763a8e6f6fc26","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-27 06:29:48.000000000","message":"Topic set to ironic-codespell","accounts_in_message":[],"_revision_number":1},{"id":"4ab02503db44f6600aca14901d9c62faf9959a7a","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2024-01-30 20:12:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2808e5ec8ac619e23a44026e8243c53bb3eb4afc","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-01-30 20:23:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c5948234d55f9783f47715575262ea4c7ed9e51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 20:24:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa0dbfd034e2b537e0a9ce70f39e1549ff49bb2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 22:04:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cfdf941b69648fdbfc1e70420bf8063\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a78170943c44a7a5db3b8d83e1dce1 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7791c32e69a34921b62be62df98dd868 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37c6cb78a71c404fa135270a28df75fc : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96dfd8f38f540c4b16f3fbf454af688 : SUCCESS in 5m 36s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/77641fff4bfc402eb03d42abdd161985 : SUCCESS in 58m 00s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/b7d0f064a0f74148be0ac9159f2f5f69 : SUCCESS in 1h 34m 22s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/52283dad9ac945a288b20f6daf11fd03 : SUCCESS in 1h 10m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/381521c3434a43d7957131fdd4759ee2 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0a05fd4d6dc544c9f3dcaf3c5d83c4edc8178f7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 22:04:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"95f0926f4776ea1086017cec5764d15b9f1ed0d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 22:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b766f7fdd0f44654b5f0e873515e1705\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11060e60582e408da66ae731d647ec2c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7d8446de7a33a1fcbbf377aca17ff684acc8be1","revisions":{"b7d8446de7a33a1fcbbf377aca17ff684acc8be1":{"kind":"REWORK","_number":1,"created":"2024-01-26 04:26:43.000000000","uploader":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"ref":"refs/changes/87/906787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/87/906787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/906787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/906787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/87/906787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/87/906787/1"}}},"commit":{"parents":[{"commit":"778403a47be6f7aa800372abe1a9418a78214306","subject":"[codespell] Adding Tox Target for Codespell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/778403a47be6f7aa800372abe1a9418a78214306"}]}],"author":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-26 04:26:38.000000000","tz":0},"committer":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-26 04:26:38.000000000","tz":0},"subject":"[codespell] Adding CI target for Tox Codespell","message":"[codespell] Adding CI target for Tox Codespell\n\nThird in a series of commits to add Codespell to Ironic Repos. This commit adds the Tox Target to CI\n\nA future commit could potentially add a git-blame-ignore-revs file if their are lots of spelling mistakes that could clutter git blame.\n\nChange-Id: Ic7afcf11342e9a33d7fb76128a0c9746ab879dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b7d8446de7a33a1fcbbf377aca17ff684acc8be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/b7d8446de7a33a1fcbbf377aca17ff684acc8be1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
