)]}'
{"id":"openstack%2Fmetalsmith~910636","triplet_id":"openstack%2Fmetalsmith~master~I6d630eaa8057799c160deba1a27d5561baf10bc7","project":"openstack/metalsmith","branch":"master","topic":"test-ci","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-03-04 11:01:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6d630eaa8057799c160deba1a27d5561baf10bc7","subject":"[DNM] TEST CI","status":"ABANDONED","created":"2024-02-29 15:22:17.000000000","updated":"2024-03-04 11:01:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f9be6cfbf7d2ec6e9e5daaac76b964fff00e263","_number":910636,"virtual_id_number":910636,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-29 16:43:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-29 16:43:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9400835fe70f0f8528df8ce41fa1fb4eaab82cc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-02-29 15:22:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cca7911d4348be8333c7a1771955464da0f86af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 16:43:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ffdcfd011664328872d7ac768543c23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/335e9c9c3ce64a4a88d788b8da34fdd2 : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9554676309a942e0a631180092ddd29c : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa7c0b49b7aa4a36acab66789a563c69 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00a854ea105341498f7b2c00f672dd6e : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785036b99e9b4b8d84aeea2b2981bf50 : SUCCESS in 5m 58s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/5cf4d30fe778440f842ad0533dbf26aa : SUCCESS in 1h 19m 56s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/fa2b8240969d4d3a8d61a01c853f63a0 : SUCCESS in 1h 11m 38s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/75646b6c610747acb5d6347f4d274458 : SUCCESS in 50m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/501d56f5d1bc4301a28e6e172b8efdfe : SUCCESS in 4m 34s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/14596d422d3642938fed3fd52c6c46f6 : SUCCESS in 2m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f9be6cfbf7d2ec6e9e5daaac76b964fff00e263","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-03-04 11:01:00.000000000","message":"Abandoned\n\nironic gate was fixed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"988a768fc010576c13a275023b280e0521e386b2","revisions":{"988a768fc010576c13a275023b280e0521e386b2":{"kind":"REWORK","_number":1,"created":"2024-02-29 15:22:17.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/36/910636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/36/910636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/36/910636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/36/910636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/36/910636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/36/910636/1"}}},"commit":{"parents":[{"commit":"9601b89f74504c59338dfcf3baa437b7fcfc5a3b","subject":"reno: Update master for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/9601b89f74504c59338dfcf3baa437b7fcfc5a3b"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-02-29 15:22:07.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-02-29 15:22:07.000000000","tz":60},"subject":"[DNM] TEST CI","message":"[DNM] TEST CI\n\nChange-Id: I6d630eaa8057799c160deba1a27d5561baf10bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/988a768fc010576c13a275023b280e0521e386b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/988a768fc010576c13a275023b280e0521e386b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
