)]}'
{"id":"openstack%2Fmetalsmith~917738","triplet_id":"openstack%2Fmetalsmith~stable%2F2024.1~I642a1579e95dfff2188206313ea6d315b0533a5f","project":"openstack/metalsmith","branch":"stable/2024.1","topic":"bootloader-stable/2024.1","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2024-05-01 21:48:25.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I642a1579e95dfff2188206313ea6d315b0533a5f","subject":"Use bootloader element instead of grub2","status":"MERGED","created":"2024-05-01 18:56:28.000000000","updated":"2024-05-01 21:50:34.000000000","submitted":"2024-05-01 21:48:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917738-bootloader-stable/2024.1","cherry_pick_of_change":916018,"cherry_pick_of_patch_set":7,"meta_rev_id":"014df086fa201451c4796242d7e39cb3c7ad701c","_number":917738,"virtual_id_number":917738,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-01 21:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-05-01 20:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-05-01 20:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 18:56:28.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"CC"},{"updated":"2024-05-01 20:29:48.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-05-01 20:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e26ce2b314244b541d19e959e9096e48132554bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-01 18:56:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2c9e7c8623c77facc4be46c67eda8261de998605","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-01 19:14:41.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"c00fbe59a8cb3e7d5f4026ff4dd1e6c5376235a1","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-05-01 20:29:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e04e16c3166af084712a861a43399d9d4dc65ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 20:35:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab76eb8d2b8460bae1fc9fa5834c987\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b408cd87cc144978090dfe08c96a83d : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f06d471bd43d4ddc8395bb95a5afc9f0 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9aa6bc2c7c52442fbf21c6e9269fcd94 : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6dcf30bca0e46928302ef642fc92b42 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79462b58dc36467fa326c61a2481e15b : SUCCESS in 6m 05s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/521bda7cbdae4ca1b8d763575e923ee7 : SUCCESS in 1h 32m 59s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/c64a87528c0c4ac68d9fbf64730cb258 : FAILURE in 55m 46s (non-voting)\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/8c25423a0b2d4b838bfc2be219ae8252 : SUCCESS in 1h 05m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7fe414a73f3f48489bdd29ab81a740d1 : SUCCESS in 5m 23s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/04796405c2414cf59ab1d02ffb6ddff5 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"541df011c0cba9623ff663870e51bacaef3fff8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 20:36:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6bd9580787a95a792f0932f8d3bf469f0d83810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 21:48:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd8bda4c1b44463a1f56e9586ab99fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b5606b018043509c1249bbbfdef798 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd3484ec809d415abe04df224716636f : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/050d3845a2834b84b57eb730802ab6d8 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007fb223ffc04c99b0f62908005bbbbd : SUCCESS in 5m 13s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/b7716818afd445bd90adb81f995d41dc : SUCCESS in 1h 00m 26s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/4ba16b1986234fa29f1d0dbee6b8998d : SUCCESS in 44m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa0b6ddb0b8b4dea95948fc959b5f9eb : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fb7fb9a17911c617fad598ffc3047f0081568662","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 21:48:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"014df086fa201451c4796242d7e39cb3c7ad701c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 21:50:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ded12e08a74e9d8bf07da482fdb093\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776f6e800a644b1c87d48686a51e1e7f : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ebd5c98ecc92ad78e952dc7151f594dbc858058","revisions":{"7ebd5c98ecc92ad78e952dc7151f594dbc858058":{"kind":"REWORK","_number":1,"created":"2024-05-01 18:56:28.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/38/917738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/38/917738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/38/917738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/38/917738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/38/917738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/38/917738/1"}}},"commit":{"parents":[{"commit":"9601b89f74504c59338dfcf3baa437b7fcfc5a3b","subject":"reno: Update master for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/9601b89f74504c59338dfcf3baa437b7fcfc5a3b"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2024-04-16 20:16:54.000000000","tz":720},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-05-01 18:56:28.000000000","tz":0},"subject":"Use bootloader element instead of grub2","message":"Use bootloader element instead of grub2\n\nbootloader is recommended since it aims to build a valid grub config\ninstead of just installing the packages.\n\nAlso, mark legacy job as non-voting for the time being\n\nThe legacy job is broken because of a know issue with tinycore related\nto the glibc library.  Moving it to non-voting for the time being while\nwe evaluate if it\u0027s worth trying to fix it or jsut remove it.\n\nAlso, install diskimage-builder from the checked-out repo, so that\ndepends on works as expected.\n\nChange-Id: I642a1579e95dfff2188206313ea6d315b0533a5f\n(cherry picked from commit 7f42d1cadf57edf1d7f5c2ee6a2aa87820a809a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7ebd5c98ecc92ad78e952dc7151f594dbc858058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/7ebd5c98ecc92ad78e952dc7151f594dbc858058"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
