)]}'
{"id":"openstack%2Fmetalsmith~918664","triplet_id":"openstack%2Fmetalsmith~master~I033a9925fe798f329b7bb9a629da0cd0f1202683","project":"openstack/metalsmith","branch":"master","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2024-05-14 18:32:31.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I033a9925fe798f329b7bb9a629da0cd0f1202683","subject":"Error with clarity when a bad upgrade was encountered","status":"MERGED","created":"2024-05-08 14:40:26.000000000","updated":"2024-05-14 18:33:31.000000000","submitted":"2024-05-14 18:32:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918664","meta_rev_id":"35a6fcb8969089aa9b21950977561d282133557d","_number":918664,"virtual_id_number":918664,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 18:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2024-05-09 22:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 17:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-14 17:22:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 16:35:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-09 22:52:39.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2024-05-14 17:22:11.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"719dccf1b446eb30492d957f0451ac41c0fda1b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-08 14:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e18584de6e3575b6a57a0a0bccb410d8720b8467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 16:35:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49ade85069b54deba70196f125c7b119\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ac98c63c644d62a624b635187dcc6a : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef394097d2bd47b698ca35b61af93900 : FAILURE in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cefd67b16de84681b7c0d1dc27a7b95d : FAILURE in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3cdc31ff9d943eca572019765b79346 : FAILURE in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc2534961b649ec8ff07289377fd5be : SUCCESS in 7m 17s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/a09a9794c14e4c49bc4f383bf16e9cc3 : SUCCESS in 1h 49m 53s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/7e9256337eaa420c872c92b916a2671f : FAILURE in 56m 36s (non-voting)\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/c218edc158cd4d3a9042fb813e674bd9 : SUCCESS in 1h 15m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8aeaf0815ba54a658118f0f567fb7ade : SUCCESS in 5m 36s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/2f4164e874d947d7b52def8fe7c34ad6 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"295049a0c3b13ae8707ef25bfb802ac3eade502b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-08 20:43:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fe259dd841152859af6df08cd0d70af1f15469a6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-08 21:03:30.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1cb46329a1c1fbf200e997932e6d7641d339dd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 22:33:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20d27d52d644976a1e18f2c517aa1c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264a4e292c60404780da68fe9c1b2123 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2121dd5e20c425b97da2311c166473d : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/076467f3549b4da8b4f4eb105f512c5b : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad869dc77fe444159ea3b151a76679f9 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e62b70d574e444db7a215fbab60bb1f : SUCCESS in 6m 38s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/9ff700ec8646486b9f35997871487f5b : SUCCESS in 1h 44m 31s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/f69d92d162814076a8f3e203448d74da : FAILURE in 56m 06s (non-voting)\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/f1b096904e5249bdb3607fdab753098f : SUCCESS in 1h 05m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1bb4a1aaf0db4fc983a924385358d79f : SUCCESS in 6m 27s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/7a697ff8ecb8432e968f6acabfd9707e : SUCCESS in 2m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f07cb93b84a82954f2da53a94679678a2a85ba80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-09 16:49:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3a2697d901a8cbef1a603cb799575a6016076585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 18:51:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380fa3ef40df48b68d97acf10b52bb06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54207ffd2544974801a25493bfcf6f2 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc65be83d216426aa34f8bad6c404ebf : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0eaea294958640ab98b283012d68fde6 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/498b83afe24f47c6b3fd636484a7477f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0c4bd4306c4b85a4eb1f5b2860adf2 : SUCCESS in 6m 59s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/3ba8dc78583347ecbc2a4bc4a409cc5f : SUCCESS in 1h 54m 47s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/5082b684db9a4e219869e962098201b3 : FAILURE in 54m 08s (non-voting)\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/d136c59c3598400c852b79e8761f46b6 : SUCCESS in 59m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5c8fd147efd426784e01ee5c4a5599a : SUCCESS in 5m 37s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/ea2741bcdd124927a090a152bff3c17e : SUCCESS in 2m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"42affe7843869f6285e511fd7fa4494644e3113e","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2024-05-09 22:52:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32df1800becce01a4eb65bfde8c958b0de52b530","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-05-14 15:46:57.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":3},{"id":"80d82c9bc4fb1362ec02ce96230f6fdf3dd892a2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2024-05-14 17:22:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05e97b7236e5ba8ae264282eb1e4aabc5a866934","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 17:22:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d429e7fd89e54ec0aa1801a90342d0797d7ede61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 18:32:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24eb8bea59d14ff0a81a685188b2ac9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2156e32daf6e4e2d879096563acb0c70 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da549d3fda164e7ba5576874f2aa4c24 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a27901243a4d499da0fd17287cca46cc : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270f3db5ac9b40979a9a421cb9c776ed : SUCCESS in 6m 27s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/1a1a277cdf4740889f4e2a820bfdf995 : SUCCESS in 1h 08m 40s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/7d0a0ae2fe8a4dbaaef5d4dc664e70a8 : SUCCESS in 46m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a9cdd2baa1fe47e8a4b3955624193947 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"17c6fa3c49e17c26e323715d34d24b2c55bf8e34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 18:32:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"35a6fcb8969089aa9b21950977561d282133557d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 18:33:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6af6dff3d094da3b4602a428332dc69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bde99f49114002aab4fe76ceb2e225 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dd5dc4e471576e05b0bd42c79508f51f6bc1b4e1","revisions":{"267789b6ac78243b0ffd2cafe3dfb8699ce7e9cf":{"kind":"REWORK","_number":1,"created":"2024-05-08 14:40:26.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/64/918664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/64/918664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/1"}}},"commit":{"parents":[{"commit":"cbb0f359d013b4c62b03fc371e94c07f76d3ceec","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cbb0f359d013b4c62b03fc371e94c07f76d3ceec"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-05-08 14:40:11.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-05-08 14:40:11.000000000","tz":-420},"subject":"DNM/WIP: Create missing allocation on list?","message":"DNM/WIP: Create missing allocation on list?\n\nChange-Id: I033a9925fe798f329b7bb9a629da0cd0f1202683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/267789b6ac78243b0ffd2cafe3dfb8699ce7e9cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/267789b6ac78243b0ffd2cafe3dfb8699ce7e9cf"}]},"branch":"refs/heads/master"},"8e993289f8293f6a1676d950266d544246dddc5e":{"kind":"REWORK","_number":2,"created":"2024-05-08 20:43:50.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/64/918664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/64/918664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/2"}}},"commit":{"parents":[{"commit":"cbb0f359d013b4c62b03fc371e94c07f76d3ceec","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cbb0f359d013b4c62b03fc371e94c07f76d3ceec"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-05-08 14:40:11.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-05-08 20:31:57.000000000","tz":-420},"subject":"Create missing allocation on list","message":"Create missing allocation on list\n\nA fun issue can occur where \"metalsmith list\" returns an error\nindicating \"Request requires an ID but none was found\".\n\nWhat is happening is metalsmith list is collecting data from\nmultiple locations, and it has identified continuty error where\na node is deployed with an instance_id field having been set,\nbut it expect and did not find an allocation record for that\nnode.\n\nThis is a fairly fatal error, or at least was since it does not\nneed to be.\n\nThe allocation can be backfilled for metalsmith not to error,\nand the list operation will do that for the user instead of\nreturning a hard error. If a hard error is still encountered,\nthen and only then is the error returned.\n\nChange-Id: I033a9925fe798f329b7bb9a629da0cd0f1202683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/8e993289f8293f6a1676d950266d544246dddc5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/8e993289f8293f6a1676d950266d544246dddc5e"}]},"branch":"refs/heads/master"},"dd5dc4e471576e05b0bd42c79508f51f6bc1b4e1":{"kind":"REWORK","_number":3,"created":"2024-05-09 16:49:50.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/64/918664/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/64/918664/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/64/918664/3"}}},"commit":{"parents":[{"commit":"cbb0f359d013b4c62b03fc371e94c07f76d3ceec","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cbb0f359d013b4c62b03fc371e94c07f76d3ceec"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-05-08 14:40:11.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2024-05-09 16:41:47.000000000","tz":-420},"subject":"Error with clarity when a bad upgrade was encountered","message":"Error with clarity when a bad upgrade was encountered\n\nMetalsmith is generally used as a frontend for baremetal\nprovisioning such that a user can ask for baremetal to be\ndeployed with a simplified command line. The challenge is\nit artificially hydrates together data from several queries\nto return a list of instances which have been deployed\nor are managed using metalsmith. The challenge is if someone\ndidn\u0027t create an allocation record, or if someone used the\nmetalsmith modules to do so when the node was in a maintenance\nstate, for example if BMC communication was not working.\n\nThe underlying challenge is this results in a lack of an allocation\nrecord from being created. This can be manually corrected, and instead\nof providing a vague error, metalsmith now provides a detailed error\nwith the command to remedy the error state through manual creation of\nthe allocation record, which allows \"metalsmith list\" to work as\nexpected.\n\nChange-Id: I033a9925fe798f329b7bb9a629da0cd0f1202683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/dd5dc4e471576e05b0bd42c79508f51f6bc1b4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/dd5dc4e471576e05b0bd42c79508f51f6bc1b4e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
