)]}'
{"id":"openstack%2Fmetalsmith~933264","triplet_id":"openstack%2Fmetalsmith~master~I10e4a3e60d1fa03ef462cdb9674bf5db873455fb","project":"openstack/metalsmith","branch":"master","topic":"ironic-fix-python-job","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-28 12:41:32.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I10e4a3e60d1fa03ef462cdb9674bf5db873455fb","subject":"Replace outdated python job template","status":"MERGED","created":"2024-10-24 11:46:19.000000000","updated":"2024-10-28 12:43:42.000000000","submitted":"2024-10-28 12:41:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933264-ironic-fix-python-job","meta_rev_id":"34adfb7331f3a05f385dad4862bd51766a663e68","_number":933264,"virtual_id_number":933264,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-28 12:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2024-10-24 16:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2024-10-28 11:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2024-10-28 11:23:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 13:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-24 16:55:34.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2024-10-28 11:23:48.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"83c9668eddc519b27447264775e587a4b7d7091f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 11:46:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e122fa4dd90ef8ec6a2c867a96bb44fc9b86da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 13:14:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba96b5e2118049e69fa0469953172d18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d00a15689d1c4f168b40baef30ea623d : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb7ac21691394903b972461cc91d9d48 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/876dee957c9a4e73869db26663348a43 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b81ceec96c54f31aa669a1ac4bf0007 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625b63749c5740a581f7c4123253fc96 : SUCCESS in 7m 20s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/a4411061d244487397658dbf20c2f42e : SUCCESS in 1h 27m 17s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/992a7675f55248099200539c5142dc70 : SUCCESS in 1h 20m 27s (non-voting)\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/bad2ba02c1104a75b57ae7b2856b2226 : SUCCESS in 56m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/07290257fffd4238980f89a6f278d4e0 : SUCCESS in 5m 25s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/7f9a7780f82c4214bc852a1a137f6e98 : SUCCESS in 2m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"426a1a9cd2396cbca6c3391b74bc34edf12a702f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 15:35:21.000000000","message":"Topic ironic-fix-python-jobs removed","accounts_in_message":[],"_revision_number":1},{"id":"a88a93091cbdbdcc23222eb69b6f1b56fadcc193","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 15:35:30.000000000","message":"Topic set to ironic-fix-python-job","accounts_in_message":[],"_revision_number":1},{"id":"1ba12574ae715f9d5b448a4f85976e0473833a20","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-24 16:55:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"047af761d82baeaa724d6f7b83fc8c48db73f59d","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-24 16:55:42.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"c400e16c275f64c24b5ed8b906665cc5da37a7b5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-10-28 11:23:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"980476edac9b31d87072d62e856dab8e8bc4ca07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 11:24:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5f56ef52bf919330b16eca61bc9454794a2e768","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:41:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3e2635ca7f407899344eb2d5f03137\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16461016aec1436d8576d7574e96097d : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a5c55a50892424fbc66d84fbe311a4d : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ca488a35dff4a758518934b40d6d227 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0948993728254d449b541671b93c30f5 : SUCCESS in 6m 34s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/a361818da48243e0b3495406a0defc3a : SUCCESS in 1h 13m 19s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/0bcce3e787b748bc8a95a5e1e4020d07 : SUCCESS in 50m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/271a69df61194cad91bf01ce9af28265 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3c7c54d2c0b2b6c518b5409ef76195f406f4adb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:41:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"34adfb7331f3a05f385dad4862bd51766a663e68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 12:43:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05cca085403a47bf8a2ac5a5e04527ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2aebae3d29442887bfac5e5fc53fa7 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ef23f36ef026ceeafbd13d9171dfa79cafc1b1b","revisions":{"2ef23f36ef026ceeafbd13d9171dfa79cafc1b1b":{"kind":"REWORK","_number":1,"created":"2024-10-24 11:46:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/64/933264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/64/933264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/933264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/933264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/64/933264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/64/933264/1"}}},"commit":{"parents":[{"commit":"c02d2f8c73f9046accb9b96ff731c69e3a4c3064","subject":"Drop SETUPTOOLS_USE_DISTUTILS\u003dstdlib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/c02d2f8c73f9046accb9b96ff731c69e3a4c3064"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 11:45:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 11:45:24.000000000","tz":540},"subject":"Replace outdated python job template","message":"Replace outdated python job template\n\nWe should use non-versioned template now, so that python versions are\nselected according to the release version.\n\nChange-Id: I10e4a3e60d1fa03ef462cdb9674bf5db873455fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2ef23f36ef026ceeafbd13d9171dfa79cafc1b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/2ef23f36ef026ceeafbd13d9171dfa79cafc1b1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
