)]}'
{"id":"openstack%2Fmetalsmith~936103","triplet_id":"openstack%2Fmetalsmith~master~Ib696cd6770010eb13755bf4e144e19f410a5e5db","project":"openstack/metalsmith","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-25 01:00:13.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ib696cd6770010eb13755bf4e144e19f410a5e5db","subject":"Remove ceilometer service overrides","status":"MERGED","created":"2024-11-24 15:38:33.000000000","updated":"2024-11-25 03:07:01.000000000","submitted":"2024-11-25 03:04:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936103","meta_rev_id":"7eebde6b6aa098192b27d6bd8ad6d82af891064d","_number":936103,"virtual_id_number":936103,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 03:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2024-11-24 16:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2024-11-24 16:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-24 16:48:36.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2024-11-24 17:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98796130f2d781ba894d2482a8dc17351ce06579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-24 15:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f94108f8f8fb5a9080fb10aa57eaed65f0a5d9c6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-11-24 16:48:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e02a93f03c81238df0055f5cd62efff46e80a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 17:19:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7487862873f54b2382556b91bf9a3a63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19b0173d9114796831dd7018bf53c09 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67939fd32af242579fb34320f0792796 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b5a77edeeb5405f9a4a6585daa9d04c : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e5f876dfdcb45f482b1842cfa292719 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1cc366e29d4a6285084b1c8ce30978 : SUCCESS in 5m 04s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/d645f71fe12c4c8e98a5529401cb2f7f : SUCCESS in 55m 41s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/87cd792c01ba48ac857530add820a537 : SUCCESS in 1h 32m 20s (non-voting)\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/475c4b18fa774a9caf17eea2517f64e3 : SUCCESS in 34m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/124a0d93da11423c9d778dc44766c00a : SUCCESS in 4m 25s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/b6a5ba22ad98417bbc29de8c898229aa : SUCCESS in 1m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c1595a9c7b15e36a12406ba1c809a6ed4bf06cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 17:20:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9f522aaed56844e6e77fa2f36c93c38de3d4464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 18:34:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1b8ceeaeb1b48928a9869fa0e28e1f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f750ecd872034ae8a2ccde758e4c231d : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b180c695ac6349e58960fc53c599c0f7 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ff807de4fcf474fb97e1c08b26238ea : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7055995be6456199b20e0f8dc767d0 : SUCCESS in 7m 09s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/ab03d20ee84d47959385b627755e780c : FAILURE in 1h 14m 02s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/70a3d93823ac4739876a74c3c9d6bf03 : SUCCESS in 49m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9ed8c3065c3428bb75ab7f46b4207d1 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"446e20b555c44e634c5fea570d189e457b2d7042","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-25 01:00:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47d8907ffd13842e3c714f15d22cc6f54f338560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 01:59:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562b2f52a4f74ccbb827ec9b1760ecf2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1685fe5bb92a4b61a17d07198ebdbb2a : SUCCESS in 2m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3ebf76f00da41e4b39a3ec5f34ffb6f : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0703b67aec5f44849a80152eb0739f81 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9079948b81474367943b5ab071f6096a : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a202f2d40e451c866b1d45cc7be0e8 : SUCCESS in 4m 03s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/b58d05cb4f78463c907a736bed03adca : SUCCESS in 58m 18s\n- metalsmith-integration-glance-centos9-legacy https://zuul.opendev.org/t/openstack/build/4cd934b945144fb18ae41562bb31e587 : SUCCESS in 42m 44s (non-voting)\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/1fa19d4807dc49628119200bc14d4432 : SUCCESS in 32m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9682cd32cde84da3803df54d748d5d28 : SUCCESS in 3m 18s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/a46a6fe2774946288d04cf6da36b0284 : SUCCESS in 2m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39d02c3949db55b9b2cc1ba9d7681fe5aa3eda70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 01:59:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20432ca893032f4a9ca1d9f2b083e3c11fc8557e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 03:04:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eaff9f8aefa4edf8008296484eb5ade\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a8519d2fdd4ba6a8c75c0efe33287b : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da0a6fb85694b40ae2b104d29bb4c78 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b569f9e4244a4528a3f23938a34bbc25 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e26829d87b843048e995e4d21a9336e : SUCCESS in 4m 15s\n- metalsmith-integration-glance-centos9-uefi https://zuul.opendev.org/t/openstack/build/fd47505bd0764bf1a4969b690c450df8 : SUCCESS in 1h 04m 01s\n- metalsmith-integration-http-cirros https://zuul.opendev.org/t/openstack/build/9fff98dd770a4eecbf0e7b9d610feab0 : SUCCESS in 33m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54d7ed390181420695b65f4125b77505 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"288f5aad6cbd169c163a24cc1879c16afbba9e0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 03:04:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7eebde6b6aa098192b27d6bd8ad6d82af891064d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 03:07:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61648ac845014bd7911ba0dd6035a41f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b93a0f64b74551b510588142f499bc : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0754ee73cea9d068a00ca6f9fe63c14576a730e9","revisions":{"0754ee73cea9d068a00ca6f9fe63c14576a730e9":{"kind":"REWORK","_number":1,"created":"2024-11-24 15:38:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/936103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/03/936103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/936103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/936103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/03/936103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/03/936103/1"}}},"commit":{"parents":[{"commit":"e992e41e1c179f949e22e9e23b67a4c40fed74f4","subject":"Update gerrit URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/e992e41e1c179f949e22e9e23b67a4c40fed74f4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:36:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-24 15:38:00.000000000","tz":540},"subject":"Remove ceilometer service overrides","message":"Remove ceilometer service overrides\n\ndevstack no longer installs ceilometer services unlesss ceilometer\ndevstack plugin is enabled, so we can safely remove the options to\ndisable ceilometer services.\n\nThis allows us to remove references to removed services such as\nceilometer-api .\n\nChange-Id: Ib696cd6770010eb13755bf4e144e19f410a5e5db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0754ee73cea9d068a00ca6f9fe63c14576a730e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/0754ee73cea9d068a00ca6f9fe63c14576a730e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
