)]}'
{"id":"openstack%2Fmetalsmith~936686","triplet_id":"openstack%2Fmetalsmith~master~I32dce2336b7f95d018d42acb9b14317f772492c7","project":"openstack/metalsmith","branch":"master","topic":"reno-eom-2023.1","hashtags":[],"change_id":"I32dce2336b7f95d018d42acb9b14317f772492c7","subject":"reno: Update master for unmaintained/2023.1","status":"MERGED","created":"2024-11-29 07:54:49.000000000","updated":"2024-11-29 08:33:41.000000000","submitted":"2024-11-29 08:32:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936686-reno-eom-2023.1","meta_rev_id":"9e24c3db748594c3e85b2aa488ee66d61127d0b4","_number":936686,"virtual_id_number":936686,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-29 08:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-11-29 08:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2024-11-29 08:18:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-29 08:15:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-29 08:18:33.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"f9dea783c960e60b0d901ee19f34cda31c31997c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-11-29 07:54:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4fdb0b472cf55ea5798cf1a81e9ed2a287693e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:15:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0446a7acebfd4dcf865bd3319fecee1a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9caf2eeb544d4d4eb61c10585aacc2a7 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb87da5681dc4ff4b8c4cc7d3233476a : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdf50b2a7d9d48a093872f28678ae719 : SUCCESS in 4m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/88a586d9e7ec4753b26909f911201800 : SUCCESS in 7m 04s\n- metalsmith-tox-codespell https://zuul.opendev.org/t/openstack/build/a7948178bb5c44bcb17ef191a2ef3d2d : SUCCESS in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf11cdbd3139719707e33453f47fea6343a936e8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-11-29 08:18:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eed6d1b20a38eb5e7d9c40bc4dc047ad9e7fcbe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:18:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f24b6df2cc3349be773704793d192383f1a13db4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:32:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9d9b548bd114a38a49f30165c6f1014\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c143c3161174984bb3f6b458b722d82 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f25ed15c114363b9e054f50a5714b1 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fca310aa0ee54439b36cc1a5b6ff0c74 : SUCCESS in 3m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3db3d5ba2a6e4c4d927a392b491f0cb7 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"189725056639dce9bc461bdf690ada9391e7fd0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:32:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e24c3db748594c3e85b2aa488ee66d61127d0b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:33:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34f0b16aa0fc4c57b28ebb5207ffdf44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe12c0928af4bc787395c419e68ca95 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36021862146042de89f50f1b8c10b265 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfcd79d3ae703b6264d14308c2c215d8d144b2f9","revisions":{"cfcd79d3ae703b6264d14308c2c215d8d144b2f9":{"kind":"REWORK","_number":1,"created":"2024-11-29 07:54:49.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/86/936686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/metalsmith","ref":"refs/changes/86/936686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/936686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/936686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/metalsmith refs/changes/86/936686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/metalsmith refs/changes/86/936686/1"}}},"commit":{"parents":[{"commit":"54a51a48fb8847f2fd23630a5eeeb44ae3549df6","subject":"Merge \"Remove ceilometer service overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/54a51a48fb8847f2fd23630a5eeeb44ae3549df6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-29 07:54:49.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-29 07:54:49.000000000","tz":0},"subject":"reno: Update master for unmaintained/2023.1","message":"reno: Update master for unmaintained/2023.1\n\nUpdate the 2023.1 release notes configuration to build from\nunmaintained/2023.1.\n\nChange-Id: I32dce2336b7f95d018d42acb9b14317f772492c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cfcd79d3ae703b6264d14308c2c215d8d144b2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/metalsmith/commit/cfcd79d3ae703b6264d14308c2c215d8d144b2f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
