)]}'
{"id":"openstack%2Fmistral-dashboard~660010","triplet_id":"openstack%2Fmistral-dashboard~master~I14ab5578cd6d4e027833f798871bd68e67627e5e","project":"openstack/mistral-dashboard","branch":"master","topic":"Bug#modify_url","hashtags":[],"change_id":"I14ab5578cd6d4e027833f798871bd68e67627e5e","subject":"Change action_execution to action_executions in url","status":"MERGED","created":"2019-05-20 03:40:51.000000000","updated":"2019-05-20 04:15:04.000000000","submitted":"2019-05-20 04:15:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660010-1558325704235-795c03e7","meta_rev_id":"3f9fa58a77d584abd5762fef1eb63bede2fc0412","_number":660010,"virtual_id_number":660010,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-20 04:15:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-05-20 04:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-05-20 04:05:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 04:05:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-20 04:15:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da0e4c37901c61bdd3c985db66d612ed23e968b1","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-05-20 03:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d618c328a880840ac61035c4801b1f1805b4976","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:48:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/660010/1/check/openstack-tox-pep8/1500987/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/10/660010/1/check/openstack-tox-py27/7fd24f2/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/10/660010/1/check/openstack-tox-py36/ccbc2eb/ : SUCCESS in 6m 11s\n- openstack-tox-py37 http://logs.openstack.org/10/660010/1/check/openstack-tox-py37/4721eee/ : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"31aa0ea9e89aa0b68f49a0028a6924d056d226f3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-20 04:05:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fd45d39c34bfce9c464e390d94b243e49c16668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:05:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47b02375885d37f5e89f2b89538498e7af2d49da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:15:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4bb4ddfab88dc27db0edefe26788c84844eab519","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:15:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/660010/1/gate/openstack-tox-pep8/ad5be70/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/10/660010/1/gate/openstack-tox-py27/bf8546c/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/10/660010/1/gate/openstack-tox-py36/9401733/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/10/660010/1/gate/openstack-tox-py37/c3794e9/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dc4afb3b91d0f191bcbbfe26de857f9da3542cb","revisions":{"7dc4afb3b91d0f191bcbbfe26de857f9da3542cb":{"kind":"REWORK","_number":1,"created":"2019-05-20 03:40:51.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/10/660010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-dashboard","ref":"refs/changes/10/660010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/10/660010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/10/660010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/10/660010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-dashboard refs/changes/10/660010/1"}}},"commit":{"parents":[{"commit":"399b9dff683fd0255fd80dc1fdb0b76653060ca9","subject":"Update sphinx dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/399b9dff683fd0255fd80dc1fdb0b76653060ca9"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-20 03:22:21.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-05-20 03:22:36.000000000","tz":480},"subject":"Change action_execution to action_executions in url","message":"Change action_execution to action_executions in url\n\nChange-Id: I14ab5578cd6d4e027833f798871bd68e67627e5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/7dc4afb3b91d0f191bcbbfe26de857f9da3542cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/7dc4afb3b91d0f191bcbbfe26de857f9da3542cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
