)]}'
{"id":"openstack%2Fmistral-dashboard~700849","triplet_id":"openstack%2Fmistral-dashboard~master~I8f97ac0e749c6c108d809f246b044ba3e8a0010d","project":"openstack/mistral-dashboard","branch":"master","topic":"drop-django111-support","hashtags":[],"change_id":"I8f97ac0e749c6c108d809f246b044ba3e8a0010d","subject":"Drop Django 1.11 support","status":"MERGED","created":"2020-01-01 20:57:34.000000000","updated":"2020-01-09 07:09:32.000000000","submitted":"2020-01-09 07:09:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700849-1578553772843-7a9322a4","meta_rev_id":"e0eb93ac4e34c6d6e197531efdae40b7963815aa","_number":700849,"virtual_id_number":700849,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 07:09:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-01-09 06:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-01-09 06:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 06:57:17.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-09 07:09:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3480ab4e0be2f21ef4b1e2494b15cf47b5f6139c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-01 20:57:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26d467ecb55557778c33bff48bfb2cfae86b38d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 21:12:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/10bb48ce88d54d12ac040536f6f7b1de : SUCCESS in 4m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3134296fe46847978819e648c6c9ec47 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781d4bdaa6c3497dbb6edd21411ab01a : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53ab6357d5874ea4951177044b9bdaa7 : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d02f1b6f89241edb1d8c634d4ddc9d6 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a33f70a7705c2a0667f8554359b8a47ff6d83fbb","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 19:29:29.000000000","message":"Topic set to drop-django111-support","accounts_in_message":[],"_revision_number":1},{"id":"7b0b9edbda82896bde7577124b010e4f34da54b9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-09 06:57:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"890e18653ad57331fc64aea1bead263a8ee8e746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 06:57:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6835d33b75b57dbb5dc9f726dae8a7f7f8c17d00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 07:09:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e0eb93ac4e34c6d6e197531efdae40b7963815aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 07:09:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/d912df31d65c4690ae517898e0509ed9 : SUCCESS in 5m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7897bfec31644d6b4fc419fc46de074 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a82d2a15da464485c8facc36bb135d : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ace3b5549fd4737ac89bd35ec4044f2 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77f05b817fc044c493add4b7627ec81d : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01dd28e53b08fdfe8749f83f29febd59387bb539","revisions":{"01dd28e53b08fdfe8749f83f29febd59387bb539":{"kind":"REWORK","_number":1,"created":"2020-01-01 20:57:34.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/49/700849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-dashboard","ref":"refs/changes/49/700849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/49/700849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/49/700849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/49/700849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-dashboard refs/changes/49/700849/1"}}},"commit":{"parents":[{"commit":"dcd4303c4b7ddd249f1b15db735f28482d3e944c","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/dcd4303c4b7ddd249f1b15db735f28482d3e944c"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-01 20:57:06.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-01 20:57:31.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\n\nDepends-On: https://review.opendev.org/#/c/700733/\nChange-Id: I8f97ac0e749c6c108d809f246b044ba3e8a0010d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/01dd28e53b08fdfe8749f83f29febd59387bb539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/01dd28e53b08fdfe8749f83f29febd59387bb539"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
