)]}'
{"id":"openstack%2Fmistral-dashboard~997809","triplet_id":"openstack%2Fmistral-dashboard~stable%2F2025.1~Ia3ff24c76fd54b6fd3931bf63c91c6f5c69381f3","project":"openstack/mistral-dashboard","branch":"stable/2025.1","topic":"pagination-objects-stable/2026.1-stable/2025.2-stable/2025.1","attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-08-11 14:32:27.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-18 12:19:06.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ia3ff24c76fd54b6fd3931bf63c91c6f5c69381f3","subject":"Add pagination for objects","status":"NEW","created":"2026-07-18 02:26:53.000000000","updated":"2026-08-11 14:32:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":997808,"cherry_pick_of_patch_set":1,"meta_rev_id":"3875dd22c1cb1397ce6ad844db2266d8086f8494","_number":997809,"virtual_id_number":997809,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-18 12:28:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-18 02:26:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2026-07-18 03:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-18 12:02:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"e1bfdf5e2c64720c43c33b5ad28ebe654c5051c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-18 02:26:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"5477c8692d7b64fc130ae01cac078404cbbfe860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 03:30:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0424ea4b4b854a8f941affc7c0c9b4c1\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30264878f3e648d3aacbc947e93541e6 : FAILURE in 2m 27s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5bd858af79a24d0094788bc1c8d63182 : RETRY_LIMIT in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57945c8b482a417d880c0bea804d98c2 : FAILURE in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3ee934cc2d84dc58d56bf98a015dcba : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29bd2d5a4d524f1aae0fa71e3a7e80ac : FAILURE in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba99e317c8545e0b388b3e473fba79c : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b576999e4d84dc50f54e5d1675b77c1005832b03","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-18 04:04:43.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"dad361ebd3a642edeb1f1122ac643ea8d952c6a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-18 12:02:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5de9e7fd0a13395ad6863673076d22c8c9cd3fa0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-18 12:02:54.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"908a07267a936677f68235455f8fceb73155d730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 12:13:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9eab567f04d54718a994eb7a62ea61b2\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da1bec1ad8e14354b3df1504593834df : FAILURE in 1m 38s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/cf94cf8408184b52a21f843de99b233e : RETRY_LIMIT in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21a64856962423cb50e8ffed75ffb8f : FAILURE in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cd450c45b6e4d928fd1a57af25fe0a6 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b98f15ffb654fbcbd201e9b869fb1fe : SUCCESS in 2m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e28a0e8833a1498fb55beffbca7c235a : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"75b9ccb8fb442aa3b537d969d5c64de333c67d2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-18 12:19:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a50f6f5a096c39ee6e7950a454264cd6e4e6cf96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 12:28:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96385471d9a74e0b88a4b26d10adccba\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/693da3804eb24059b0a09734d2e6f80d : SUCCESS in 4m 00s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/687bbf03f4e241738e6cffda2a58e209 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528fed79474448d7a0dc8edeaf0074df : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5642addbecfb4574b11660abccac06b0 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19d8a222b75445294dca4ae64df83a2 : SUCCESS in 2m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8baf10543a2e42f4b1eea11eca7f83a4 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"49a74f21248e5ba00c34b4faa24d6dba5150fd3c","revisions":{"161249d919a7d8361b634d0d64da8b399cdebbd6":{"kind":"REWORK","_number":1,"created":"2026-07-18 02:26:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/997809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-dashboard","ref":"refs/changes/09/997809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/1"}}},"commit":{"parents":[{"commit":"5174872afee81572d0ce98a56cf0a9049ba084e2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/5174872afee81572d0ce98a56cf0a9049ba084e2"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-11 12:09:00.000000000","tz":420},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-18 02:26:53.000000000","tz":0},"subject":"Add pagination for objects","message":"Add pagination for objects\n\nChange-Id: Ia3ff24c76fd54b6fd3931bf63c91c6f5c69381f3\nSigned-off-by: Bo Tran \u003cministry.96.nd@gmail.com\u003e\n(cherry picked from commit 1b6681cabe2fd6765ef96051b64cc5b971caf041)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/161249d919a7d8361b634d0d64da8b399cdebbd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/161249d919a7d8361b634d0d64da8b399cdebbd6"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"98abef7043ec3c119c45669dde1b2a3b7429c154","ours":"5174872afee81572d0ce98a56cf0a9049ba084e2","theirs":"40137f4e45b332e87b3294e809ea6572cbc40a12","merge_strategy":"recursive","contains_conflicts":false}},"a677d445af6f730a97810b3e103a6852b174c0b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-18 12:02:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/997809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-dashboard","ref":"refs/changes/09/997809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/2"}}},"commit":{"parents":[{"commit":"20c1954f97e20dee1557b73f9d8fcdd4562711d7","subject":"Constrain setuptools\u003c82 due to pkg_resources deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/20c1954f97e20dee1557b73f9d8fcdd4562711d7"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-11 12:09:00.000000000","tz":420},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-18 12:02:48.000000000","tz":0},"subject":"Add pagination for objects","message":"Add pagination for objects\n\nChange-Id: Ia3ff24c76fd54b6fd3931bf63c91c6f5c69381f3\nSigned-off-by: Bo Tran \u003cministry.96.nd@gmail.com\u003e\n(cherry picked from commit 1b6681cabe2fd6765ef96051b64cc5b971caf041)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/a677d445af6f730a97810b3e103a6852b174c0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/a677d445af6f730a97810b3e103a6852b174c0b2"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase","conflicts":{"base":"5174872afee81572d0ce98a56cf0a9049ba084e2","ours":"161249d919a7d8361b634d0d64da8b399cdebbd6","theirs":"20c1954f97e20dee1557b73f9d8fcdd4562711d7","merge_strategy":"recursive","contains_conflicts":false}},"49a74f21248e5ba00c34b4faa24d6dba5150fd3c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-18 12:19:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/997809/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-dashboard","ref":"refs/changes/09/997809/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-dashboard refs/changes/09/997809/3"}}},"commit":{"parents":[{"commit":"982fecb91dc722986b930895d35040f56a12839b","subject":"Constrain setuptools\u003c82 due to pkg_resources deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/982fecb91dc722986b930895d35040f56a12839b"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-11 12:09:00.000000000","tz":420},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-18 12:19:00.000000000","tz":0},"subject":"Add pagination for objects","message":"Add pagination for objects\n\nChange-Id: Ia3ff24c76fd54b6fd3931bf63c91c6f5c69381f3\nSigned-off-by: Bo Tran \u003cministry.96.nd@gmail.com\u003e\n(cherry picked from commit 1b6681cabe2fd6765ef96051b64cc5b971caf041)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/49a74f21248e5ba00c34b4faa24d6dba5150fd3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-dashboard/commit/49a74f21248e5ba00c34b4faa24d6dba5150fd3c"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase","conflicts":{"base":"20c1954f97e20dee1557b73f9d8fcdd4562711d7","ours":"a677d445af6f730a97810b3e103a6852b174c0b2","theirs":"982fecb91dc722986b930895d35040f56a12839b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
