)]}'
{"id":"openstack%2Fmistral-extra~690969","triplet_id":"openstack%2Fmistral-extra~master~I218efe01ad56dbe04666716ae61ea2d97ef6912d","project":"openstack/mistral-extra","branch":"master","hashtags":[],"change_id":"I218efe01ad56dbe04666716ae61ea2d97ef6912d","subject":"tox: Keeping going with docs","status":"MERGED","created":"2019-10-24 13:10:31.000000000","updated":"2019-11-05 07:25:40.000000000","submitted":"2019-11-05 07:25:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690969-1572938740591-8bdcbf74","meta_rev_id":"ec293f7d611d87d85549fa57fe4943bd0e5e6fb8","_number":690969,"virtual_id_number":690969,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 07:25:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-11-05 07:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-03 08:28:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-11-05 07:17:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-03 08:28:18.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-05 07:17:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-05 07:25:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a80b1aa32651382cd12ca4fe9b380ef69ac15b6","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-24 13:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"274b410c64ca7c084359b8ff1eb8491622d11889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:20:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d8fb2dae9f412e962ba32303248e78 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70bbafd5debb48ad82924363bbef1363 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faeffdf102054a279c526e37d464540c : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc520090ac8e4fadbe7758181ce944b0 : SUCCESS in 5m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b68f22e75cf4c5eb7c4f8a82982b592 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ec04fb315181a5cf1d4446a981fe41da8637e7d8","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-24 13:22:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44d9c131776f81b60fd633ae62e11ccd7156d73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:33:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea8b198bbe94b15871e0b9e621f68d5 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32a346de18274c0ea8865243f10b0253 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fefdd3dd3d64d1eafcad4ef5f76d775 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/099a60d872f44fbcaf1f43a0ea117507 : SUCCESS in 3m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/530b464db665431fa169ef1246293595 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"6ac28dc520360e1cc00f0c019db79b24733e38d1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-03 08:28:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"51b8cc895908a76cc416f0a304f5ae3c59c4ded6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-05 07:11:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ac30fe1827e4e7b0682ab89765eddb272d4e7efe","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-05 07:17:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c3e1dc2771571680fda3a7f8ea68210d9e908d8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 07:17:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e61527211dec81a522dec9b9cceb143b21099bf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 07:25:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ec293f7d611d87d85549fa57fe4943bd0e5e6fb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 07:25:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce1ae700bba4260b37c27eb9e32936d : SUCCESS in 2m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5253fc116f4433785680016530fd139 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f9c79dc5a03449e89afcd882bd89090 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3434f9dfc97e4ed88c7fd53f8450c325 : SUCCESS in 3m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8219a8e0355d4a0aa1e232f8ba1982f7 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"078205550c9492cdb2d1b9fd85bf9e717af3780e","revisions":{"9bb03ca57ed75f408ccb52ae19009067eaf0cd59":{"kind":"REWORK","_number":1,"created":"2019-10-24 13:10:31.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/69/690969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/69/690969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/1"}}},"commit":{"parents":[{"commit":"0a68426bd8513ffdecec3d2b893a8ab09e5b09a6","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/0a68426bd8513ffdecec3d2b893a8ab09e5b09a6"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 13:10:08.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 13:10:27.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I218efe01ad56dbe04666716ae61ea2d97ef6912d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/9bb03ca57ed75f408ccb52ae19009067eaf0cd59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/9bb03ca57ed75f408ccb52ae19009067eaf0cd59"}]},"branch":"refs/heads/master"},"078205550c9492cdb2d1b9fd85bf9e717af3780e":{"kind":"REWORK","_number":2,"created":"2019-10-24 13:22:30.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/69/690969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/69/690969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/69/690969/2"}}},"commit":{"parents":[{"commit":"0a68426bd8513ffdecec3d2b893a8ab09e5b09a6","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/0a68426bd8513ffdecec3d2b893a8ab09e5b09a6"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 13:10:08.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 13:22:06.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nSphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nChange-Id: I218efe01ad56dbe04666716ae61ea2d97ef6912d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/078205550c9492cdb2d1b9fd85bf9e717af3780e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/078205550c9492cdb2d1b9fd85bf9e717af3780e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
