)]}'
{"id":"openstack%2Fmistral-extra~694564","triplet_id":"openstack%2Fmistral-extra~master~I32edd618a65f5ef65be1c7894dd4f52c97f733c3","project":"openstack/mistral-extra","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I32edd618a65f5ef65be1c7894dd4f52c97f733c3","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-15 14:59:27.000000000","updated":"2019-11-21 16:11:59.000000000","submitted":"2019-11-21 16:11:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694564-1574352719111-9f2e12eb","meta_rev_id":"967503e609a755319ccde4f316896372b13fca4c","_number":694564,"virtual_id_number":694564,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 16:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-21 15:50:09.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-11-18 04:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-11-17 13:13:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-11-18 04:36:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-17 13:13:56.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-18 04:36:17.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-21 16:11:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd057d4b7dffc7c210fc733faffdc5a01b1eaf6b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 14:59:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a48a8df6ec840cb200c3e0fb1fef82ecc71304b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 15:07:40.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"99b55ad5b1774718266f89d1574d1b9781d22c7e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-16 17:01:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"676c3f97c78f7d66821c1127eb59a508975596d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 17:08:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6dfec7a5736447419913dcd412a733bb : SUCCESS in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd99da01b4f4a1680de4780bf0f4215 : SUCCESS in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68b6fc94e2d5484ab809a03c6922be55 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abf2084f40a4426ebec318b49a86e568 : SUCCESS in 3m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f4aa2fd175d47a5a9ed6f27b9d84ce2 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e94ca95ef9276487d6807ed59caef49d22fa8f46","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-17 13:13:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0e8545939b27c7fb377b9218a63384b766c2dab9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-18 04:36:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"050d8327c3440f8909bf95b57f0febd88f226061","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 15:50:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4e51433f7dbc5ab5a3c3f6e4c03a35d3d5e88df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:01:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc79d2e9a7f24304bc23bcd172c2d9c4 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cfe5ea7222c4e6194e712d6b28f8e46 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14f2314deef745b7b46ae97ddecb4dcd : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b80f4af10fb4854abb3f36ddfdc6bae : SUCCESS in 3m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dfdcb55f3c48490db088f5c0abffd274 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":2},{"id":"23012e0b369a495dc0d4ad6c53f23b300bb3894b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:01:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86aa760bd0dd1a09371afa7631b53098452db318","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:11:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/62ed568f85a440a3b03fbf743f3db631 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c8719c705f4e0e804992d6b8923e73 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35aef6313978446f860d743dc45addc6 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79f03c13ed57409db4a20e53d01b1bdf : SUCCESS in 3m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/764fb3c84b82446a91bfb63e59d624a8 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fb2caa3cec266fc869ec6b7ecc77d81e95e35aee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:11:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a99070183b0be61dfb0564300461f4a9a699f0e","revisions":{"47d916914df25e3177772f4858c46752fb3e42cf":{"kind":"REWORK","_number":1,"created":"2019-11-15 14:59:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/694564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/64/694564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/1"}}},"commit":{"parents":[{"commit":"33ff45b5eae71f93155ef9a8fc1260f813f08da7","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/33ff45b5eae71f93155ef9a8fc1260f813f08da7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 14:55:46.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 14:59:24.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nmistral-extra is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: I32edd618a65f5ef65be1c7894dd4f52c97f733c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/47d916914df25e3177772f4858c46752fb3e42cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/47d916914df25e3177772f4858c46752fb3e42cf"}]},"branch":"refs/heads/master"},"1a99070183b0be61dfb0564300461f4a9a699f0e":{"kind":"REWORK","_number":2,"created":"2019-11-16 17:01:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/694564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/64/694564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/64/694564/2"}}},"commit":{"parents":[{"commit":"33ff45b5eae71f93155ef9a8fc1260f813f08da7","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/33ff45b5eae71f93155ef9a8fc1260f813f08da7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 14:55:46.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 17:01:53.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nmistral-extra is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: I32edd618a65f5ef65be1c7894dd4f52c97f733c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/1a99070183b0be61dfb0564300461f4a9a699f0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/1a99070183b0be61dfb0564300461f4a9a699f0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
