)]}'
{"id":"openstack%2Fmistral-extra~705676","triplet_id":"openstack%2Fmistral-extra~master~I93cc4d3d28fda2ec29fec3e828dbed92cbb24683","project":"openstack/mistral-extra","branch":"master","hashtags":[],"change_id":"I93cc4d3d28fda2ec29fec3e828dbed92cbb24683","subject":"Correct the mistral-extra package name","status":"MERGED","created":"2020-02-04 13:33:06.000000000","updated":"2020-02-05 04:20:08.000000000","submitted":"2020-02-05 04:20:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705676-1580876408470-4edda447","meta_rev_id":"1d8417cc55ae24c94f8d0bcce84b6d160a24c58d","_number":705676,"virtual_id_number":705676,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 04:20:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":2,"date":"2020-02-04 13:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2020-02-05 04:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2020-02-05 04:11:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 13:44:54.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-02-05 04:11:58.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-05 04:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0e11d38f53d3149f1344eeb2cb1668ac78069bf","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-02-04 13:33:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a49919d4ba03b682247bf31ac7d844a9204058c2","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-02-04 13:44:54.000000000","message":"Patch Set 1: Code-Review+2\n\noops. Good catch","accounts_in_message":[],"_revision_number":1},{"id":"0b54c19b761c1f326a3c5f60507a9213c3de26fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 13:45:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97a0761d4c1b48c9a2886232596515fe : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6146d9177ca74a8a8e632d1fbfe54abb : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d242d0e3c3f4647af6ab3a2c9b7c2bc : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bcf578e33984659808881804acda4cb : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67a1ab41211b498598cd03e772e3a06a : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6da54f106d142a5afcce6ad19e86de8 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"00c683e027b1675ff18c20ba0ab2d2758d4f7799","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-05 04:11:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2404f0eef427a4527bef52a11c55582914018acf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 04:12:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d066aa4936df8a2899b1ce8b0b377f3290cc4a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 04:20:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83e917c8d1884726815d78607b24586e : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cfa856ecd6443fbad7caadf0bcda993 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63bad5dee28a457e818fa95881978f0c : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecd28cbdab994956ab59c854a39ee22e : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/964098b81f9e445eaa99de43b4f87e5b : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1d8417cc55ae24c94f8d0bcce84b6d160a24c58d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 04:20:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c158891ac532eb5df6949baea92aa9e886eb014e","revisions":{"c158891ac532eb5df6949baea92aa9e886eb014e":{"kind":"REWORK","_number":1,"created":"2020-02-04 13:33:06.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/76/705676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/76/705676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/76/705676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/76/705676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/76/705676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/76/705676/1"}}},"commit":{"parents":[{"commit":"57afe0a3f46e7ea86db1b04e59b0d78f4d9af99f","subject":"Fix requirements, tox , zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/57afe0a3f46e7ea86db1b04e59b0d78f4d9af99f"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-04 13:32:28.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-04 13:33:03.000000000","tz":0},"subject":"Correct the mistral-extra package name","message":"Correct the mistral-extra package name\n\nThe project name is mistral-extra, but the Python package is\nmistral_extra.\n\nChange-Id: I93cc4d3d28fda2ec29fec3e828dbed92cbb24683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/c158891ac532eb5df6949baea92aa9e886eb014e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/c158891ac532eb5df6949baea92aa9e886eb014e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
