)]}'
{"id":"openstack%2Fmistral-extra~706475","triplet_id":"openstack%2Fmistral-extra~master~Ie9897c612ec16985ecd0b1c2ee0b058a8df781d9","project":"openstack/mistral-extra","branch":"master","hashtags":[],"change_id":"Ie9897c612ec16985ecd0b1c2ee0b058a8df781d9","subject":"Don\u0027t fail the unit tests if a client isn\u0027t installed","status":"ABANDONED","created":"2020-02-07 11:45:03.000000000","updated":"2020-02-10 12:08:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3462687f6082fb1044ed3a2d472209807d420664","_number":706475,"virtual_id_number":706475,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-07 11:59:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":2,"date":"2020-02-07 13:24:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-02-09 13:59:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-02-07 13:12:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-02-07 13:12:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"date":"2020-02-07 13:12:30.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2020-02-07 13:12:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 11:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-07 13:12:30.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-02-07 13:12:30.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-07 13:12:30.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-02-07 13:12:30.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-02-07 13:24:02.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-02-09 13:59:11.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"abf4fc3e1b415fcccacf4e3d2f276fdc610a6f8a","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-02-07 11:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96d7fe87237dfa1a7a795194536a2683f07904a7","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-02-07 11:45:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b41dbc6c86075c418d5d89d8a0a2aaeaa4d7b314","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-02-07 11:50:51.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a48ee207748d712acc805d2cce9de63f9bece4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 11:59:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27efa0c327b4247b96f0eb8e8c013a0 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/510d2c23a11e423a8016438e1b55018e : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/195df16ad9964872852eafbd9c5c7536 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7620d9b4069411fb88bfece0f0da16f : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b415524a7c47448cbbceef318bbf9015 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4edac6e77d49938ba96e6f86aed175 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e4f92d767e67f966639d6ab4d52c00db4873f5fe","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-02-07 13:24:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"15f2f91866001017f0beddd245048a6a7c2b420e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-09 13:59:11.000000000","message":"Patch Set 3: Code-Review-1\n\nI would use the mock method which was done to designate and baremetal inspector clients in the setUp method\n\nWhat I don\u0027t understand is how it worked in mistral. The same unit test runs on mistral Do you run the unit tests when packaging mistral rpm ?","accounts_in_message":[],"_revision_number":3},{"id":"139b7c7606b6ee5d40839c10905a0472c1d69725","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-09 14:34:40.000000000","message":"Patch Set 3:\n\n\u003e I would use the mock method which was done to designate and\n \u003e baremetal inspector clients in the setUp method\n \u003e \n \u003e What I don\u0027t understand is how it worked in mistral. The same unit\n \u003e test runs on mistral Do you run the unit tests when packaging\n \u003e mistral rpm ?\n\nI would mock the zun client which doesn\u0027t have an rpm packaging if it fails to import","accounts_in_message":[],"_revision_number":3},{"id":"6d21ea97ea6a30aac357dbbd4dd0ab4161ee9f3a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-10 08:17:33.000000000","message":"Patch Set 3:\n\nI did a fix here https://review.opendev.org/#/c/706745/1","accounts_in_message":[],"_revision_number":3},{"id":"3462687f6082fb1044ed3a2d472209807d420664","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-02-10 12:08:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"770fc4cc3122140ac793eec5b3413c304c7c8bf0","revisions":{"904dbbc1c6bb65ec3d3e37ad1b3cd36b0ecc87a6":{"kind":"REWORK","_number":1,"created":"2020-02-07 11:45:03.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/75/706475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/75/706475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/1"}}},"commit":{"parents":[{"commit":"91ff80c79d765f9d08fc9556526b289a3c18072d","subject":"Merge \"Move openstack actions to mistral-extra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/91ff80c79d765f9d08fc9556526b289a3c18072d"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-07 11:41:00.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-07 11:45:01.000000000","tz":0},"subject":"Don\u0027t fail the unit tests if a client isn\u0027t installed","message":"Don\u0027t fail the unit tests if a client isn\u0027t installed\n\nThe Python clients should be considered optional. This is why we don\u0027t\nerror when importing them. This changes the test_generator test to be\ntollerant of clients that are not installed.\n\nThis can happen in a couple of situations;\n- we don\u0027t want to fail if one client is bad and doesn\u0027t install correctly.\n- Alternatively when packaged by downstream consumers (such as the RDO rpms)\n  not all OpenStack clients are packaged. they will then be missing but the\n  package building process still uses the tests for verification.\n\nChange-Id: Ie9897c612ec16985ecd0b1c2ee0b058a8df781d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/904dbbc1c6bb65ec3d3e37ad1b3cd36b0ecc87a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/904dbbc1c6bb65ec3d3e37ad1b3cd36b0ecc87a6"}]},"branch":"refs/heads/master"},"0551fc94c8b9743df0ef03593cc7f7bd59955c70":{"kind":"REWORK","_number":2,"created":"2020-02-07 11:45:29.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/75/706475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/75/706475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/2"}}},"commit":{"parents":[{"commit":"91ff80c79d765f9d08fc9556526b289a3c18072d","subject":"Merge \"Move openstack actions to mistral-extra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/91ff80c79d765f9d08fc9556526b289a3c18072d"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-07 11:41:00.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-07 11:45:21.000000000","tz":0},"subject":"Don\u0027t fail the unit tests if a client isn\u0027t installed","message":"Don\u0027t fail the unit tests if a client isn\u0027t installed\n\nThe Python clients should be considered optional. This is why we don\u0027t\nerror when importing them. This changes the test_generator test to be\ntollerant of clients that are not installed.\n\nThis can happen in a couple of situations;\n- we don\u0027t want to fail if one client is bad and doesn\u0027t install correctly.\n- Alternatively when packaged by downstream consumers (such as the RDO rpms)\n  not all OpenStack clients are packaged. they will then be missing but the\n  package building process still uses the tests for verification.\n\nChange-Id: Ie9897c612ec16985ecd0b1c2ee0b058a8df781d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/0551fc94c8b9743df0ef03593cc7f7bd59955c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/0551fc94c8b9743df0ef03593cc7f7bd59955c70"}]},"branch":"refs/heads/master"},"770fc4cc3122140ac793eec5b3413c304c7c8bf0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-07 11:50:51.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/75/706475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/75/706475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/75/706475/3"}}},"commit":{"parents":[{"commit":"91ff80c79d765f9d08fc9556526b289a3c18072d","subject":"Merge \"Move openstack actions to mistral-extra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/91ff80c79d765f9d08fc9556526b289a3c18072d"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-07 11:41:00.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2020-02-07 11:50:31.000000000","tz":0},"subject":"Don\u0027t fail the unit tests if a client isn\u0027t installed","message":"Don\u0027t fail the unit tests if a client isn\u0027t installed\n\nThe Python clients should be considered optional. This is why we don\u0027t\nerror when importing them. This changes the test_generator test to be\ntollerant of clients that are not installed.\n\nThis can happen in a couple of situations;\n- we don\u0027t want to fail if one client is bad and doesn\u0027t install correctly.\n- Alternatively when packaged by downstream consumers (such as the RDO rpms)\n  not all OpenStack clients are packaged. they will then be missing but the\n  package building process still uses the tests for verification.\n\nNeeded-By: https://review.rdoproject.org/r/#/c/14216/\nChange-Id: Ie9897c612ec16985ecd0b1c2ee0b058a8df781d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/770fc4cc3122140ac793eec5b3413c304c7c8bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/770fc4cc3122140ac793eec5b3413c304c7c8bf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
