)]}'
{"id":"openstack%2Fmistral-extra~894345","triplet_id":"openstack%2Fmistral-extra~master~I90086707346d549cb4181199b194020d9c931c25","project":"openstack/mistral-extra","branch":"master","topic":"reno-2023.2","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-23 09:46:02.000000000","reason":"Change was submitted"},"35978":{"account":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"last_update":"2024-09-23 09:46:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90086707346d549cb4181199b194020d9c931c25","subject":"Update master for stable/2023.2","status":"MERGED","created":"2023-09-08 14:55:12.000000000","updated":"2024-09-23 09:47:37.000000000","submitted":"2024-09-23 09:46:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894345-reno-2023.2","meta_rev_id":"a4e2a66a330991fdd24352f046aa756014db2a99","_number":894345,"virtual_id_number":894345,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-23 09:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-23 09:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-23 09:34:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 15:11:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-12 09:09:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-09-19 14:41:09.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"state":"REVIEWER"},{"updated":"2024-09-23 09:34:41.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"aa1bb512c708a3d6ba5ed5462621961532e0e57c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-08 14:55:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fc2821bdcc6320162685776338226d175869eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:11:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fef298532ef48c29277ceb1cbbe7bb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b5735fa2bc415d871db80655d3cd35 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed969c1571f641728c60680fb5155fe7 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5497d0bd5b4eaaae665090df2f8bb8 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"eada9694b01ae6606b85a2352eec1e8a1829153b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-12 09:09:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e602c6d47948aabeb848959e6359b1a2c98157e","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-19 14:41:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"113c54384b7b2b2779d4dd268f5aa027e0853a16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-09-23 08:11:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2b352389d3030a553912acb92b159b9d616aea0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 08:17:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d6f58a1a87a432fbea0eabba9bb961c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ca511a7412413396053cd5ede15df9 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/119a7bd339f04a0cbf358beccd9ae6db : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aec700afa37145c18c4d763d8c8bc6ed : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c8a366e6b3731b7b58f19207e3dd7d3b0041fa70","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-09-23 09:34:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6ab79e95415182c1bd91ca79e0fceb0c80c45c9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 09:35:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d33af40904669faa3b9234e57792298eef2349ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 09:46:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c2d9499cb68471bb9af5b6a29351fbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca85de914b14ed4bafb60410c83b4e7 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f7b3b01ff7470a8683694420886346 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/648b31af3e624706b3cdb088c87eefde : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"26c800ecd33a80e43082ac2c36fe679564ba7fe3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 09:46:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a4e2a66a330991fdd24352f046aa756014db2a99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 09:47:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae071e1b49d43b999af68a32333b593\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/183ee1dc0ba3416d81a05642cc13ff79 : SUCCESS in 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a4ff84e4cf24fe7b2f625529eab5b00 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a79c34829df6edc9538919948dbdcf1b62f4c8b","revisions":{"a2684829987a4ca5cfcfd153b56ef07eec929a8e":{"kind":"REWORK","_number":1,"created":"2023-09-08 14:55:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/45/894345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/45/894345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/1"}}},"commit":{"parents":[{"commit":"2e0139b9b2ef9d23f889c148a641d0ea3b631102","subject":"Merge \"Add monitoring server to mistral-extra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/2e0139b9b2ef9d23f889c148a641d0ea3b631102"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-08 14:55:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-08 14:55:09.000000000","tz":0},"subject":"Update master for stable/2023.2","message":"Update master for stable/2023.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2023.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2023.2.\n\nSem-Ver: feature\nChange-Id: I90086707346d549cb4181199b194020d9c931c25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/a2684829987a4ca5cfcfd153b56ef07eec929a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/a2684829987a4ca5cfcfd153b56ef07eec929a8e"}]},"branch":"refs/heads/master"},"2a79c34829df6edc9538919948dbdcf1b62f4c8b":{"kind":"REWORK","_number":2,"created":"2024-09-23 08:11:17.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/45/894345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/45/894345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/45/894345/2"}}},"commit":{"parents":[{"commit":"f6e6aeee205f59b683b71a01801e1236dea90765","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/f6e6aeee205f59b683b71a01801e1236dea90765"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-08 14:55:09.000000000","tz":0},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-09-23 08:09:56.000000000","tz":120},"subject":"Update master for stable/2023.2","message":"Update master for stable/2023.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2023.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2023.2.\n\nSem-Ver: feature\nChange-Id: I90086707346d549cb4181199b194020d9c931c25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/2a79c34829df6edc9538919948dbdcf1b62f4c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/2a79c34829df6edc9538919948dbdcf1b62f4c8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
