)]}'
{"id":"openstack%2Fmistral-extra~936792","triplet_id":"openstack%2Fmistral-extra~master~I8bb0d3a6eea04901f3f54fbaf8ef8eee7edadc57","project":"openstack/mistral-extra","branch":"master","hashtags":[],"change_id":"I8bb0d3a6eea04901f3f54fbaf8ef8eee7edadc57","subject":"Get endpoint with service_name","status":"MERGED","created":"2024-11-30 23:01:42.000000000","updated":"2024-12-01 09:56:19.000000000","submitted":"2024-12-01 09:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936792","meta_rev_id":"dbc3e1bd46e080f4a4a785bc8de2045208cf5ca5","_number":936792,"virtual_id_number":936792,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-01 09:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-01 09:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-01 09:45:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-30 23:05:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de99503e402ea6065be8c7ad2ff5e16801360cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-11-30 23:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"563824b933627a5e98f83a52ac6a57a17496bbb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 23:05:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af6ed3ba4302409e87297ee596cfd265\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c2b90c477e845188f7b7c7d59759756 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f069397cf144c04a662c1b556995163 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2030d4a5c0640e18bfd2af2c64534c2 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6d2853fa404a5585f3db23d9e986a5 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2b7da5dee53b9aab122d4bdce12b5cd1764ec5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-11-30 23:08:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b42d9f45e375bf0daa38ecd215e870df43230eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 23:12:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13789768a6434b108936779f4da897f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c176d45347fc4e4587aa27aaa5350c02 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/050ace79987a4e68a41fa9cf9d1cb51b : SUCCESS in 2m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5e6bd36ed5c4bf6b0321bf727fd2a03 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5069392976d942f1a7ad8142cb7a4f42 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4a967afd3c76a62adb88b1e5514aec18b8ebf543","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-01 09:45:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b1d0a31e840ffdf0308dc1129da6f3511899cadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-01 09:45:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfdc4ead64903b5946869597bd0e769c8b7b4acd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-01 09:55:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06391c6b76d94a3abed3ee4fc6ea4fe2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70cd888259344851a8acd973f4f779a6 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecb5a62ea8e0492db35a5156a7cfc2fd : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aec70a8b972d49f1845d1e8f6f742f47 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3dd5cdccfc34847baee5af40f097933 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b4e3290a16ef83d30e4f90501610f6cfd593fde4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-01 09:55:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dbc3e1bd46e080f4a4a785bc8de2045208cf5ca5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-01 09:56:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/515a53f55a6a49b3aa5bf24741a8ecb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e276a4c816ac41ef86dd95689a7ddff0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46b60f2ffb10f3de45872de22242bf3ce8266d2c","revisions":{"406df27ebcc79728b4a153b06b7f02e82ab6002e":{"kind":"REWORK","_number":1,"created":"2024-11-30 23:01:42.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/92/936792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/92/936792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/1"}}},"commit":{"parents":[{"commit":"6c8492d4ee341136db10c3873fc2605a44feb9fd","subject":"Merge \"Pin hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/6c8492d4ee341136db10c3873fc2605a44feb9fd"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-11-30 22:59:09.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-11-30 23:01:32.000000000","tz":60},"subject":"Get endpoint with service_name","message":"Get endpoint with service_name\n\nWe recently introduced a change to handle multiple service types.\nThis is preventing some services to be retrieved from name if the\nservice type is empty (like swift).\n\nSo, let\u0027s do a first loop over service types, then try to get endpoint\nwith service name as fallback.\n\nRelated-bug: #2090618\n\nChange-Id: I8bb0d3a6eea04901f3f54fbaf8ef8eee7edadc57\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/406df27ebcc79728b4a153b06b7f02e82ab6002e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/406df27ebcc79728b4a153b06b7f02e82ab6002e"}]},"branch":"refs/heads/master"},"46b60f2ffb10f3de45872de22242bf3ce8266d2c":{"kind":"REWORK","_number":2,"created":"2024-11-30 23:08:42.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/92/936792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/92/936792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/92/936792/2"}}},"commit":{"parents":[{"commit":"6c8492d4ee341136db10c3873fc2605a44feb9fd","subject":"Merge \"Pin hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/6c8492d4ee341136db10c3873fc2605a44feb9fd"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-11-30 22:59:09.000000000","tz":60},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2024-11-30 23:08:33.000000000","tz":60},"subject":"Get endpoint with service_name","message":"Get endpoint with service_name\n\nWe recently introduced a change to handle multiple service types.\nThis is preventing some services to be retrieved from name if the\nservice type is empty (like swift).\n\nSo, let\u0027s do a first loop over service types, then try to get endpoint\nwith service name as fallback.\n\nRelated-bug: #2090618\n\nChange-Id: I8bb0d3a6eea04901f3f54fbaf8ef8eee7edadc57\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/46b60f2ffb10f3de45872de22242bf3ce8266d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/46b60f2ffb10f3de45872de22242bf3ce8266d2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
