)]}'
{"id":"openstack%2Fmistral-extra~952029","triplet_id":"openstack%2Fmistral-extra~master~I9b5162d1d7de896b59f057fb80b3ae8459951967","project":"openstack/mistral-extra","branch":"master","topic":"license-classifier","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-06-09 07:38:36.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-06-09 07:34:56.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I9b5162d1d7de896b59f057fb80b3ae8459951967","subject":"Replace license classifier","status":"MERGED","created":"2025-06-08 13:23:55.000000000","updated":"2025-06-09 07:39:18.000000000","submitted":"2025-06-09 07:38:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952029-license-classifier","meta_rev_id":"6eaf9beb870ad50631678169d6c19f0956e1a9fb","_number":952029,"virtual_id_number":952029,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 07:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-09 07:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-09 07:34:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-08 13:24:58.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-06-08 13:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3f5503063a9d38ce80cb046b1dcdfeffb1092eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:23:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e6152971c41b38deda59819335711fd23710ae9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:24:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c6c56f582dfd22d967a112f64fa6084ad63a4ed7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:24:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a9bb9591e13dddaccd2f3f3f2f4c6b763aa69208","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:24:58.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"b5fc52a4080433a1ed3f34f2e2b620a517191870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 13:43:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ef6da739648494faaf037d1a52ad6a8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ca9b9f5572bd49eaabeae4e779e52290 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d3468f6b8741f89a6a38172da54818 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccce7d9097654efebd3e90110f47b1b6 : SUCCESS in 1m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e20dbcec21047199bbb5c91027f1617 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba51a3a84ac84f6883f3ea19b4a08659 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d202f636c34550a1c0263690a6944b : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"77453d1c24b9ab68fcc1e127f26320d0e0d29b08","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-06-09 07:34:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"020e45f0580049add734d48d93c1a803328532aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:35:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aa89ba942fb0fc62819d721abde370621ce17b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:38:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72e3b331c3834298b94ebd9f55d4bfd7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4ea887ed3b254b9a9415c7e27cb760e4 : SUCCESS in 1m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ecf0e97c9f4102b2d14a20d04c5394 : SUCCESS in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13bee2be5c3d464c914b63a5658dfdc6 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e433b5d61a044d948a3f432110c1a6a4 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc17e658caf34f308031dacbcb823b68 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"fbc4260953903d9b0448575454ad715b4b3fd536","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:38:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6eaf9beb870ad50631678169d6c19f0956e1a9fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:39:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ea8e2e4667411bb76b20027a643063\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b58c16d2bd4d868d15ef5285d6c202 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3a2aeeb1201aaf71f47053b82b1c031414302255","revisions":{"51bbdbf54f336ea2a7915700a0d7e19112e5eebc":{"kind":"REWORK","_number":1,"created":"2025-06-08 13:23:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/952029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/29/952029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/1"}}},"commit":{"parents":[{"commit":"baf142aea9bea12e02f5d1cd22b455730624fed6","subject":"Merge \"Replace pkg_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/baf142aea9bea12e02f5d1cd22b455730624fed6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:23:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:23:55.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I9b5162d1d7de896b59f057fb80b3ae8459951967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/51bbdbf54f336ea2a7915700a0d7e19112e5eebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/51bbdbf54f336ea2a7915700a0d7e19112e5eebc"}]},"branch":"refs/heads/master"},"b318aa83f68784c3f27c42a96b5a8114b79ff46d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-08 13:24:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/952029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/29/952029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/2"}}},"commit":{"parents":[{"commit":"baf142aea9bea12e02f5d1cd22b455730624fed6","subject":"Merge \"Replace pkg_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/baf142aea9bea12e02f5d1cd22b455730624fed6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:23:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:24:17.000000000","tz":0},"subject":"Replace license classifier","message":"Replace license classifier\n\n... according to the following warning from setuptools.\n\nSetuptoolsDeprecationWarning: License classifiers are deprecated.\n\n********************************************************************************\nPlease consider removing the following classifiers in favor of a SPDX license expression:\n\nLicense :: OSI Approved :: Apache Software License\n\nSee https://packaging.python.org/en/latest/guides/writing-pyproject-toml/#license for details.\n********************************************************************************\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I9b5162d1d7de896b59f057fb80b3ae8459951967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/b318aa83f68784c3f27c42a96b5a8114b79ff46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/b318aa83f68784c3f27c42a96b5a8114b79ff46d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3a2aeeb1201aaf71f47053b82b1c031414302255":{"kind":"REWORK","_number":3,"created":"2025-06-08 13:24:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/29/952029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-extra","ref":"refs/changes/29/952029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-extra refs/changes/29/952029/3"}}},"commit":{"parents":[{"commit":"baf142aea9bea12e02f5d1cd22b455730624fed6","subject":"Merge \"Replace pkg_resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/baf142aea9bea12e02f5d1cd22b455730624fed6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:23:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:24:51.000000000","tz":0},"subject":"Replace license classifier","message":"Replace license classifier\n\n... according to the following warning from setuptools.\n\nSetuptoolsDeprecationWarning: License classifiers are deprecated.\n\n********************************************************************************\nPlease consider removing the following classifiers in favor of a SPDX license expression:\n\nLicense :: OSI Approved :: Apache Software License\n\nSee https://packaging.python.org/en/latest/guides/writing-pyproject-toml/#license for details.\n********************************************************************************\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I9b5162d1d7de896b59f057fb80b3ae8459951967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/3a2aeeb1201aaf71f47053b82b1c031414302255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-extra/commit/3a2aeeb1201aaf71f47053b82b1c031414302255"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
