)]}'
{"id":"openstack%2Fmistral-lib~692975","triplet_id":"openstack%2Fmistral-lib~stable%2Frocky~I544d3c172f2dea02c62c49c311c4b5954413ae15","project":"openstack/mistral-lib","branch":"stable/rocky","topic":"bug/1850843","hashtags":[],"change_id":"I544d3c172f2dea02c62c49c311c4b5954413ae15","subject":"Ensure we mask sensitive data from Mistral Action logs","status":"MERGED","created":"2019-11-05 10:43:22.000000000","updated":"2020-06-17 13:28:18.000000000","submitted":"2019-11-06 04:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692975-1573015147925-df6c426c","meta_rev_id":"f4ae56adaa4cc8b063c7f40e9f5321739f09b45b","_number":692975,"virtual_id_number":692975,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-06 04:39:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2019-11-06 04:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2019-11-06 04:30:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 10:43:22.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-11-06 04:30:11.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-06 04:39:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"429db744d22d13171651eacf0b96fc0724a02a86","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-05 10:43:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ae1d07f3e321c1f5f8cbe13e54be94e3c70770","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-05 10:50:19.000000000","message":"Topic set to bug/1850843","accounts_in_message":[],"_revision_number":1},{"id":"a0427cadbe7ea60ecc6be98c511e0f19f82be7be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:06:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088b9e30bd684476be15a101eb5387eb : FAILURE in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9042cda5b75d459e9f371520a4b5cebf : FAILURE in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2310705ce130404a99788194405d6466 : FAILURE in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9bf96fb6f0041ef935aa640324cb9e3 : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8c31c856ace17edb6db4b3c4aac536e707c0c9d0","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-05 11:15:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c40388a3f9e7afba5388c3fcfa0d953686076156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:26:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45ad285ce364a0299d489da3e385ad6 : FAILURE in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4663222d6ef5473db93beadc2a22badf : SUCCESS in 3m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93cc36e1c5834d3daef2af78fc4e6e93 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a07832750f82425da1fbd2f280c540c2 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3836183510d92e23379cd3d19916292f15c8a508","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-05 12:30:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dbd0119c9f3ef813ae62baf53acad037eadd8499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 12:46:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c619ef1bcb3a4cb89eb2930e4fd32550 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16a98eafc37e4a288308ba745a2f62c4 : SUCCESS in 3m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/34ae36035b46483687274453d6d075ab : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89d4d23470d148088bd627d06ea286db : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c36740de2e04f371fd0ede0a446a2a24220a8249","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-06 04:30:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"56ed24de8bce05e0162eb548f3158d020e11d62f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 04:30:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"364d1f5ca86e877aa575ddf5b6a92077a78f4679","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 04:39:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f2e4b6b68458fba1cbbeb7bd7fea504375d698cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 04:39:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6345e7e3eba4f5797d23dc9320b050d : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/851e1f6daf3a42fca46babe03ad56c12 : SUCCESS in 2m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/528ae07789c14e54ac581a732f912778 : SUCCESS in 2m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a34d310e462e49f294df9f02dfcd178e : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f4ae56adaa4cc8b063c7f40e9f5321739f09b45b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-06-17 13:28:18.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit abf520d2f3de4b07541b1426656085be1eb47570","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1265d9d53aa01117e337fa1799d005113041a169","revisions":{"92ccb64263f34fdb7db068f2db106f7386f76173":{"kind":"REWORK","_number":1,"created":"2019-11-05 10:43:22.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/75/692975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/75/692975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/1"}}},"commit":{"parents":[{"commit":"e49a29ac49a082633d30e2281ce39cc23e80b75b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/e49a29ac49a082633d30e2281ce39cc23e80b75b"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-11-01 10:47:35.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-05 10:43:10.000000000","tz":0},"subject":"Ensure we mask sensitive data from Mistral Action logs","message":"Ensure we mask sensitive data from Mistral Action logs\n\nMistral didn\u0027t make use of the oslo_utils \"mask_password\" methods,\nleading in sensitive data leakage in its logs.\n\nThis patch corrects this security issue.\nNote that it depends on oslo_utils patch adding new patterns, and\nensuring it\u0027s case-insensitive.\n\nChange-Id: I544d3c172f2dea02c62c49c311c4b5954413ae15\nRelated-Bug: #1850843\nCo-Authored-By: Dougal Matthews \u003cdougal@redhat.com\u003e\nSigned-off-by: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\n(cherry picked from commit f51d2a21c0e75fe0e20990877e0615d4ace6af89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/92ccb64263f34fdb7db068f2db106f7386f76173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/92ccb64263f34fdb7db068f2db106f7386f76173"}]},"branch":"refs/heads/stable/rocky"},"aea60c6abe98eb173c9c1f30daea68e792836e51":{"kind":"REWORK","_number":2,"created":"2019-11-05 11:15:42.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/75/692975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/75/692975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/2"}}},"commit":{"parents":[{"commit":"e49a29ac49a082633d30e2281ce39cc23e80b75b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/e49a29ac49a082633d30e2281ce39cc23e80b75b"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-11-01 10:47:35.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-05 11:15:31.000000000","tz":0},"subject":"Ensure we mask sensitive data from Mistral Action logs","message":"Ensure we mask sensitive data from Mistral Action logs\n\nMistral didn\u0027t make use of the oslo_utils \"mask_password\" methods,\nleading in sensitive data leakage in its logs.\n\nThis patch corrects this security issue.\nNote that it depends on oslo_utils patch adding new patterns, and\nensuring it\u0027s case-insensitive.\n\nChange-Id: I544d3c172f2dea02c62c49c311c4b5954413ae15\nRelated-Bug: #1850843\nCo-Authored-By: Dougal Matthews \u003cdougal@redhat.com\u003e\nSigned-off-by: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\n(cherry picked from commit f51d2a21c0e75fe0e20990877e0615d4ace6af89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/aea60c6abe98eb173c9c1f30daea68e792836e51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/aea60c6abe98eb173c9c1f30daea68e792836e51"}]},"branch":"refs/heads/stable/rocky"},"1265d9d53aa01117e337fa1799d005113041a169":{"kind":"REWORK","_number":3,"created":"2019-11-05 12:30:36.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/75/692975/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/75/692975/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/75/692975/3"}}},"commit":{"parents":[{"commit":"e49a29ac49a082633d30e2281ce39cc23e80b75b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/e49a29ac49a082633d30e2281ce39cc23e80b75b"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-11-01 10:47:35.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-11-05 12:30:23.000000000","tz":0},"subject":"Ensure we mask sensitive data from Mistral Action logs","message":"Ensure we mask sensitive data from Mistral Action logs\n\nMistral didn\u0027t make use of the oslo_utils \"mask_password\" methods,\nleading in sensitive data leakage in its logs.\n\nThis patch corrects this security issue.\nNote that it depends on oslo_utils patch adding new patterns, and\nensuring it\u0027s case-insensitive.\n\nChange-Id: I544d3c172f2dea02c62c49c311c4b5954413ae15\nRelated-Bug: #1850843\nCo-Authored-By: Dougal Matthews \u003cdougal@redhat.com\u003e\nSigned-off-by: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\n(cherry picked from commit f51d2a21c0e75fe0e20990877e0615d4ace6af89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/1265d9d53aa01117e337fa1799d005113041a169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/1265d9d53aa01117e337fa1799d005113041a169"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
