)]}'
{"id":"openstack%2Fmistral-lib~705615","triplet_id":"openstack%2Fmistral-lib~master~Iaea15a982cf84fb95461b62e30aaa242d9c0097e","project":"openstack/mistral-lib","branch":"master","topic":"py3","hashtags":[],"change_id":"Iaea15a982cf84fb95461b62e30aaa242d9c0097e","subject":"Remove python2 support","status":"MERGED","created":"2020-02-04 09:03:52.000000000","updated":"2020-02-20 11:07:58.000000000","submitted":"2020-02-20 11:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705615-1582196789735-a5ae4dff","meta_rev_id":"5c76de3d3f6a11d66712f30cf28fd50fca0b663e","_number":705615,"virtual_id_number":705615,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-20 11:06:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-02-04 09:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-02-20 10:56:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 10:56:01.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-20 11:06:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2e8150572a20bf64ea74f418c075c99e74a1a67","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-04 09:03:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e0d816ed43f98a0cdd56f9d148045f2a626d78f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 09:10:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0bdb2bc317c1497dace6e63039a18605 : SUCCESS in 2m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f5ea7f713184d82939d7b8e63686ebb : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013083defe57401288e5c776dcfb72c8 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/969c9b58ec1a4ef5aa659d018f82b77e : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/816d3fd1a83c4020adfd62b961983d25 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c40e34dbffd4fae9b4608a047ab8034 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78a644bd4f940c2a0dd1dec34a2d3ce : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e1fd6352832f619c8c971eaef3f9de27af37aea1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-04 09:16:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fd24df4c46ce6bd5708a8004b7e41267bf084176","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-04 09:24:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bea2795ae74c4e726754edcfe3f1de53a98f9670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 09:25:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0558766a52fd480a86cfbfb280376b77 : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6562984b4dc74f93b263b8a29950d1de : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ea1a3bd4514ca29ad010ef82355985 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/996914eeebca42158d9b41ca13e8e255 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20dcb7c321ee432dbc697564575f6a94 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99f6717f99834099b276293cd2f0553a : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24b69580fed44d88131f7a990458ce2 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bbcf2edf3d9bd46e33305ee3b699d0ac66078d75","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-20 10:56:01.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e7d4f38b7d2d8c7a3484a094304ba2bc3d052b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 10:56:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c0a69e5e0638d756b9645ade25d83694499133f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 11:06:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7009c9d752d83ff74255d935e288e23d59fec326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 11:06:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a7d61a6389bb480e8a70693e0aa8b16e : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d887d37cfadc4bd687ea1439e339fef1 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c41824dc5c1d4fbfb949f33e373510f4 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fad51ec734641328134191477d712b7 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/706d9cff37e24caa9f0a9fa06a0bf716 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8da91cc06754d28a6ea6619068e12d6 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9d3781ff00406f93dee676b11f4d9cfeb8ec244b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 11:07:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a7fc431ac3440ba68f8a3ddaaac605 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa5e3f1d1dbf1c4b2044a8335cd03a9a66c833ff","revisions":{"6d87c73caf1119a732c5e1f46289ef11b632db93":{"kind":"REWORK","_number":1,"created":"2020-02-04 09:03:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/15/705615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/15/705615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/1"}}},"commit":{"parents":[{"commit":"1a81f2cdd85b0a48492c9b30fe1461a508e17ebc","subject":"Drop py2 support and add zuul jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/1a81f2cdd85b0a48492c9b30fe1461a508e17ebc"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-04 09:01:36.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-04 09:03:18.000000000","tz":120},"subject":"Fix requirements","message":"Fix requirements\n\nRemove python2 support\n\nChange-Id: Iaea15a982cf84fb95461b62e30aaa242d9c0097e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/6d87c73caf1119a732c5e1f46289ef11b632db93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/6d87c73caf1119a732c5e1f46289ef11b632db93"}]},"branch":"refs/heads/master"},"fa5e3f1d1dbf1c4b2044a8335cd03a9a66c833ff":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-04 09:16:50.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/15/705615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/15/705615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/15/705615/2"}}},"commit":{"parents":[{"commit":"1a81f2cdd85b0a48492c9b30fe1461a508e17ebc","subject":"Drop py2 support and add zuul jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/1a81f2cdd85b0a48492c9b30fe1461a508e17ebc"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-04 09:01:36.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-04 09:16:46.000000000","tz":0},"subject":"Remove python2 support","message":"Remove python2 support\n\nRemove support from setup.cfg\n\nChange-Id: Iaea15a982cf84fb95461b62e30aaa242d9c0097e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/fa5e3f1d1dbf1c4b2044a8335cd03a9a66c833ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/fa5e3f1d1dbf1c4b2044a8335cd03a9a66c833ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
