)]}'
{"id":"openstack%2Fmistral-lib~706001","triplet_id":"openstack%2Fmistral-lib~master~I47cc8ab5b348af28062ab3ff5329b96d2cbccdbb","project":"openstack/mistral-lib","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I47cc8ab5b348af28062ab3ff5329b96d2cbccdbb","subject":"[ussuri][goal] Cleanup python 2.7 removal","status":"MERGED","created":"2020-02-05 16:05:57.000000000","updated":"2020-02-06 07:43:55.000000000","submitted":"2020-02-06 07:41:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706001-1580974879347-99670080","meta_rev_id":"39afe9e8c55723158033c6c0ec822320672c14f9","_number":706001,"virtual_id_number":706001,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-02-06 07:26:59.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 07:41:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2020-02-06 07:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-06 07:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"date":"2020-02-06 07:23:17.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2020-02-06 07:27:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-02-06 07:23:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-02-06 07:23:17.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"date":"2020-02-06 07:23:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2020-02-06 07:23:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 07:23:17.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-02-06 07:23:17.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-06 07:23:17.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-02-06 07:23:17.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-02-06 07:23:17.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-02-06 07:26:21.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-02-06 07:27:27.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-02-06 07:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65f5e567eac7dced3c0bf28ca6ddf69a5939d57f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-05 16:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd5b754c0a6a0010cfed5f37a4a22aef93c771df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:19:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bb83a2923d8248e7905b18c735a362f7 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef845fec500b46748e5a33ca136c327f : SUCCESS in 2m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9865dea0d34d44ec83dab6289aa3713f : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f085248242418896085dcaeb2c7d77 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adb47bd63eb847c2912537867e65f393 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/196d933d9f894ebc98f7005ec29247ea : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31bdf7a106054595970c3d8538cee599 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f75c0171694dd3980df02075a274a6 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47d779a498624eebb473ae175bc28c27 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"efa2382cc51b7aa067b1ca88bc3177f1f25b4205","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-05 16:22:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21d699f46fbeaeb0c621c69513edffd8ae161a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:35:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6ed04dda14964fc3ac150dd1df61a565 : SUCCESS in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/066dd276346e4c5494b1e82f2aeb5196 : SUCCESS in 2m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/661f47fd8e774901a1cf1c3613b162bc : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c081e368904391aff4557b32f79013 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d06ab41b539a4029aef833ed8c662aa2 : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/130cb07792f1441a9d6418ea6b4ee94e : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cc2177c842d4d4e97d0f03bfd6466e3 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256e22757af2467687b42d98c73d8020 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbfe3a85b9c9440ba1cf1e3b0a92d916 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3ea0347ca7b7b5d81693912c6cfee7d9b6cd6b9b","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-06 07:24:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"87e80c855d06ca00a6576cecfa249e460b3a69fc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-06 07:26:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c1b8a4b53aa1f8dcc1ee190a9b079edc3bed3fc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-06 07:26:59.000000000","message":"Patch Set 2:\n\nWow, two +2s at the same time - thanks. Could either of you +A, please?","accounts_in_message":[],"_revision_number":2},{"id":"fc9d9fda196f49d18483a22f15e7c68e16d3f74c","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-06 07:27:27.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"835130c4c931edb7c9f8c21e9f154a5d54ced66e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:27:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2054073fc4b0e13232b29f4636db39b5dbb762e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:41:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"07ea57ab9d1401943353c254d97a630e03879f4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:41:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a540b7ee0d44377ba2d1bd03635d634 : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/348922e7f0f540e49c812eadd0b991de : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/744c3b09ebf2434fb344f44dad251837 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbb635fc0b444de9d2921ac86a7b2ec : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7824a4c807004debabdb7e0120c1c94e : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00c02cf864c9488ea3e662dca2a7828a : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f11629eef54c2b93c64422e7349953 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48fc035c11a74f6dafcdfbf84d888782 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e10650605a6e6d25ea74289772ca2f8a05e2e38b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:43:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56f748244fd4f038b05a7b3122bee48 : SUCCESS in 2m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb2b2018d5074bef99640a719a653e0c : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b97e38391881204e2ada54df8e4ec47fba33ac49","revisions":{"f5704d23882a7f508920cc272e97f2727c5eb957":{"kind":"REWORK","_number":1,"created":"2020-02-05 16:05:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/706001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/01/706001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/1"}}},"commit":{"parents":[{"commit":"1a81f2cdd85b0a48492c9b30fe1461a508e17ebc","subject":"Drop py2 support and add zuul jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/1a81f2cdd85b0a48492c9b30fe1461a508e17ebc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-05 16:03:37.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-05 16:05:49.000000000","tz":60},"subject":"[ussuri][goal] Cleanup python 2.7 removal","message":"[ussuri][goal] Cleanup python 2.7 removal\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nThis removes some cruft from Python 2.7 support.\n\nChange-Id: I47cc8ab5b348af28062ab3ff5329b96d2cbccdbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/f5704d23882a7f508920cc272e97f2727c5eb957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/f5704d23882a7f508920cc272e97f2727c5eb957"}]},"branch":"refs/heads/master"},"b97e38391881204e2ada54df8e4ec47fba33ac49":{"kind":"REWORK","_number":2,"created":"2020-02-05 16:22:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/706001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/01/706001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/01/706001/2"}}},"commit":{"parents":[{"commit":"1a81f2cdd85b0a48492c9b30fe1461a508e17ebc","subject":"Drop py2 support and add zuul jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/1a81f2cdd85b0a48492c9b30fe1461a508e17ebc"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-05 16:03:37.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-05 16:22:45.000000000","tz":60},"subject":"[ussuri][goal] Cleanup python 2.7 removal","message":"[ussuri][goal] Cleanup python 2.7 removal\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nThis removes some cruft from Python 2.7 support.\nRaise requests lower-constraint to satisfy oslo.config.\n\nChange-Id: I47cc8ab5b348af28062ab3ff5329b96d2cbccdbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/b97e38391881204e2ada54df8e4ec47fba33ac49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/b97e38391881204e2ada54df8e4ec47fba33ac49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
