)]}'
{"id":"openstack%2Fmistral-lib~716242","triplet_id":"openstack%2Fmistral-lib~master~I2fe891525bc86beb92aecf9ac2d8a490837c47d3","project":"openstack/mistral-lib","branch":"master","topic":"add_json_serialization_utils","hashtags":[],"change_id":"I2fe891525bc86beb92aecf9ac2d8a490837c47d3","subject":"Add a utility for JSON serialization","status":"MERGED","created":"2020-03-31 11:14:09.000000000","updated":"2020-04-02 06:54:57.000000000","submitted":"2020-04-02 06:52:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716242-1585810320388-cfc8ac43","meta_rev_id":"bbc5412df80ed48eeb7a5d4a488931abd429ed9d","_number":716242,"virtual_id_number":716242,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 06:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-04-02 06:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2020-03-31 14:09:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-31 11:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-04-02 06:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"date":"2020-03-31 11:40:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-04-01 04:37:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 11:40:34.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-03-31 11:44:36.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-03-31 14:09:00.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-04-01 04:37:03.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-04-02 06:41:42.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-04-02 06:52:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3220136babf8518b9447963685830d7326cbd36f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-31 11:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb2a4a1264cd5dd620f617f8234378bdcedbbbd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 11:21:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eedea1d3fd744fd2a92c2ba470d894d3 : FAILURE in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bed909b350442c5a20f4d314b0c1f7d : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1129b1a83ff944be80fb6edfa817f728 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33d06d53226f46a38830be68b4f18058 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e99d1d189e0b4cf493bbf20e42981d80 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff5ce4d8dd4a429e9a054888383eb3ff : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4fd93506d14e229266e70b8e0d92a9 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1e549ba9f55be8189992e7f18e24177fd14a898a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-31 11:33:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"040b3a5d45cd6fad3a8b2ebc84755c1b664d3f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 11:40:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fe5c99a387294f698a1cc2072c33d61f : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80dea1c03c384e48a5f2d03c0bc9ca7e : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a3601ec51c4f00b6e8465e9553bab3 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac6c22f28a2143c68996959c627b8098 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ac8021f0efb45f7b0d3506526e4b8a5 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/770032fadc68434981ebd2d2a59eaeae : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15002b7a06d64c35b47d1c78b930115a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8efd8bcc8ca90226a9e4d8db18834a9904a250ec","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-31 11:44:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"79a6e9106001de95061a58ef541ae9bdae820874","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-31 14:09:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f5f6ffb5b626afe8e4e0049aa97c84e9deb7f52b","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-04-02 06:41:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0698632346e520949aac07c076ac57905d0eb5d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 06:41:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f30adcc7a6cdaaa98ea6e547f9951524359483d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 06:52:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/65f8a41129c64205a26a3b5cf79863d0 : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1af92bd39f8b460c9b6bf25d0fe5aba1 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42083201489747b0a5d55841b2a674e2 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f5a32c83e464ca493aaaabfc3edfb1c : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df2ac2be5a6141ba9baf8771e504c5de : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c6a95b9d4c43c68b00edbfad7213fd : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"e188296c2f85da8301cd7cb1891b9e6b675886f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 06:52:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"63596d2b5c13a13ec04497c03c4683174a890edd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 06:54:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f99b84d7a1466697b18fb0fe74d2d9 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44a27384601113cdc5068940d07b85a003ed2211","revisions":{"95389810f255052491673fa48bc0afe9f1ed6f9a":{"kind":"REWORK","_number":1,"created":"2020-03-31 11:14:09.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/42/716242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/42/716242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/1"}}},"commit":{"parents":[{"commit":"237a10d22e4dc4965c553489fe3d24e8c75d308d","subject":"Support getting a file list from a module that isn\u0027t mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/237a10d22e4dc4965c553489fe3d24e8c75d308d"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-31 11:09:08.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-31 11:09:08.000000000","tz":420},"subject":"Add a utility for JSON serialization","message":"Add a utility for JSON serialization\n\n* This patch adds a utility that helps serialize data into a JSON\n  string that might contain some non-standard data types like\n  generators iterators and frozen dicts coming from YAQL. The\n  utility uses oslo.serialization project that already takes care\n  of iterators and any kinds of custom dicts. And in addition, it\n  handles generators (assuming a generator represents an iterable\n  similar to an iterator).\n* Unit tests.\n* Added YAQL into requirements and bumped the version of\n  oslo.serialization to make sure to have the \"fallback\" parameter\n  in \"jsonutils.to_primitive\"\n\nChange-Id: I2fe891525bc86beb92aecf9ac2d8a490837c47d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/95389810f255052491673fa48bc0afe9f1ed6f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/95389810f255052491673fa48bc0afe9f1ed6f9a"}]},"branch":"refs/heads/master"},"44a27384601113cdc5068940d07b85a003ed2211":{"kind":"REWORK","_number":2,"created":"2020-03-31 11:33:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/42/716242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/42/716242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/42/716242/2"}}},"commit":{"parents":[{"commit":"237a10d22e4dc4965c553489fe3d24e8c75d308d","subject":"Support getting a file list from a module that isn\u0027t mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/237a10d22e4dc4965c553489fe3d24e8c75d308d"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-31 11:09:08.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-31 11:33:05.000000000","tz":420},"subject":"Add a utility for JSON serialization","message":"Add a utility for JSON serialization\n\n* This patch adds a utility that helps serialize data into a JSON\n  string that might contain some non-standard data types like\n  generators iterators and frozen dicts coming from YAQL. The\n  utility uses oslo.serialization project that already takes care\n  of iterators and any kinds of custom dicts. And in addition, it\n  handles generators (assuming a generator represents an iterable\n  similar to an iterator).\n* Unit tests.\n* Added YAQL into requirements and bumped the version of\n  oslo.serialization to make sure to have the \"fallback\" parameter\n  in \"jsonutils.to_primitive\"\n\nChange-Id: I2fe891525bc86beb92aecf9ac2d8a490837c47d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/44a27384601113cdc5068940d07b85a003ed2211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/44a27384601113cdc5068940d07b85a003ed2211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
