)]}'
{"id":"openstack%2Fmistral-lib~719218","triplet_id":"openstack%2Fmistral-lib~stable%2Fussuri~I7fd74fd9b909e68acb84382fcc9de0c9e0af8fcd","project":"openstack/mistral-lib","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I7fd74fd9b909e68acb84382fcc9de0c9e0af8fcd","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-11 18:44:30.000000000","updated":"2020-04-13 06:58:02.000000000","submitted":"2020-04-13 06:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719218-1586760978309-332a2ac9","meta_rev_id":"b977e9b5d74503acb347a14352a034106e2e2cef","_number":719218,"virtual_id_number":719218,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 06:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-11 20:06:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-13 06:44:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-11 18:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-13 06:44:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-11 18:53:00.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-11 20:06:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-13 06:44:00.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-13 06:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14a23ab33dc0d1d99f9d7431e88287789a38e09e","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-11 18:44:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6867f0ee817a76ab6adbb7bdfaa26a1987e9ce2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-11 18:53:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67ed1526842fdd872ad116831432afe64e703d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 18:57:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf8f7864673d46b7bc3f741ca446f525 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a4159a8c0b4d108205ff73ec632ae0 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d80e6cb67fc24edeac146a2f119ec8f2 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d162bb64f40146889f5eee3480c29038 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69afa01b445a43c88d7235cefc841037 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3986f1313ca04881465c353954afce466a5c0bf1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-11 20:06:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a7ccdee071bac62d9fcae0b524f56829ac171c48","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-13 06:44:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed3e87419e52acc8185e88359c3b98dde1832bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 06:44:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"457222313f50bba2368b63c777d07931ebb0a8a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 06:56:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7fcf774b802640f707b8e68d249a4a092ee1f42a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 06:56:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/070b0893b84b4be88395722657a709fb : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8d4898842c4e5fa1d7648da7ecd07a : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bc6da8e18e3474697f7aea9d5ebcc78 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2a78be38fff41a38deb1d49c77cc3f6 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af159fa9c164cdf9be6671a9d490dcd : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3101716ba45bc9c482a2ee1ccb401c0056ef596b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 06:58:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf40e96d92824cedb076d9f5c5804a2b : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14c63eae36c0caecd19a4cf84516e2f04ae261b1","revisions":{"14c63eae36c0caecd19a4cf84516e2f04ae261b1":{"kind":"REWORK","_number":1,"created":"2020-04-11 18:44:30.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/18/719218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/18/719218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/18/719218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/18/719218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/18/719218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/18/719218/1"}}},"commit":{"parents":[{"commit":"44a27384601113cdc5068940d07b85a003ed2211","subject":"Add a utility for JSON serialization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/44a27384601113cdc5068940d07b85a003ed2211"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-11 18:44:28.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-11 18:44:28.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: I7fd74fd9b909e68acb84382fcc9de0c9e0af8fcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/14c63eae36c0caecd19a4cf84516e2f04ae261b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/14c63eae36c0caecd19a4cf84516e2f04ae261b1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
