)]}'
{"id":"openstack%2Fmistral-lib~722837","triplet_id":"openstack%2Fmistral-lib~master~I2106373bc267a780eb600eead17dc12aaddf0dd1","project":"openstack/mistral-lib","branch":"master","topic":"py38-tox","hashtags":[],"change_id":"I2106373bc267a780eb600eead17dc12aaddf0dd1","subject":"Bump default tox env from py37 to py38","status":"MERGED","created":"2020-04-24 15:28:28.000000000","updated":"2020-04-27 04:44:01.000000000","submitted":"2020-04-27 04:42:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722837-1587962571660-354c625b","meta_rev_id":"de23a0e557c4073bf032f53eecb6e0ea9fe656c7","_number":722837,"virtual_id_number":722837,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 04:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-27 04:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-24 15:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-27 04:37:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 15:31:52.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-27 04:37:04.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-27 04:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58e763efa94b06f60fbd61f7d892cd8408f67ae6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 15:28:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9979f921c0140b1227a857ad1624c25c9c6d6872","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-24 15:31:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc7fb80659c50f935da6ed7c3aa5b4415b5eb6e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 15:50:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26e68eb284bf4c1491ceef24e4fecec4 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f88879ac01ce4c089f32599f87960c22 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b34cf5a73c0148328a0205997715283f : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a674ad3eea6e49298e9221713c7f1505 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4920a37744a74aeabc0c0e8e96525771 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a64c77340eda496eb0bfa400a2f33ad3 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"63dcece8ed527b23b8e17e061eff3f2cbb661bff","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-27 04:37:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c104b871518d844d4d9a0e16ac44d54873e03ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 04:37:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2969c5a1f6e363843f7fb8ca7593cdf232dea921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 04:42:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0feb7e0d681d4c2384397e93d473d7ca : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f1d7cba77e4c5c8e0675e98617b084 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2191fafdc2744d8a099d996a146ab14 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5e7385521434e008b6caee73636366f : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7fb823d40c40f0a023528722f867d4 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/882632d36b2b4b0b9d7d6134402b4df7 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cdf768b75b2d60f24893a86d73359d51143757c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 04:42:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ecc92034217a7d6fbb30a0b94443f856866d5fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 04:44:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65efca468a34018810372c91932c3a4 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85b9f913a7bf417b91942b1a6e9915a8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32cc70f089990c89e97e93cd5d1175e3f0a2110a","revisions":{"32cc70f089990c89e97e93cd5d1175e3f0a2110a":{"kind":"REWORK","_number":1,"created":"2020-04-24 15:28:28.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/37/722837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/37/722837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/37/722837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/37/722837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/37/722837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/37/722837/1"}}},"commit":{"parents":[{"commit":"55eba920e74c29259d0d2962a21d7458b13a4436","subject":"Add py38 package metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/55eba920e74c29259d0d2962a21d7458b13a4436"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:43.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:43.000000000","tz":-300},"subject":"Bump default tox env from py37 to py38","message":"Bump default tox env from py37 to py38\n\nPython 3.8 is now our highest level supported python runtime.\nThis updates the default tox target environments to swap out\npy37 for py38 to make sure local development testing is\ncovering this version.\n\nThis does not impact zuul jobs in any way, nor prevent local\ntests against py37. It just changes the default if none is\nexplicitly provided.\n\nChange-Id: I2106373bc267a780eb600eead17dc12aaddf0dd1\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/32cc70f089990c89e97e93cd5d1175e3f0a2110a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/32cc70f089990c89e97e93cd5d1175e3f0a2110a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
