)]}'
{"id":"openstack%2Fmistral-lib~734373","triplet_id":"openstack%2Fmistral-lib~master~I7f0cc935291e5c0c4f905ad2aa0fb03e6192abb0","project":"openstack/mistral-lib","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"I7f0cc935291e5c0c4f905ad2aa0fb03e6192abb0","subject":"drop mock from lower-constraints","status":"MERGED","created":"2020-06-08 23:23:23.000000000","updated":"2020-06-09 04:02:33.000000000","submitted":"2020-06-09 04:01:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734373-1591675281495-df6bd473","meta_rev_id":"acfa68f99e2662365596db6941a9b81fd9ccab5c","_number":734373,"virtual_id_number":734373,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 04:01:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-06-09 03:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-06-09 03:53:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 03:53:19.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-06-09 04:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de32a03c9f681f5d59820d4ab03e8cd891b79e19","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 23:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09e34b0894367b2e45ee173b90a394425ebdae7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:35:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/78480cb7c6554c0db341c599532212ea : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10410727aa9d414c98d58dc2e4c2e705 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b2f4cc4fcc4feeb82512671c9ff580 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15ef8ae69de240f286e839f35c58e873 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a24b60365ea145298f2b7b04c7cce297 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2da128498a84555adcccd449083019d : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d2ec599e84893a5cfb70ff95fc1efb774850fafb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-09 03:53:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0740613d544c5da2fa5770331447cce5e015c05e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 03:53:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de3591d25f610ead86384c992a2fc9bd85f0da90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 04:01:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c8afdca0c86459789153d16da1883aa : SUCCESS in 2m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51b94731e6014b80b6c5dc3a001ddffc : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1b9c8fc33fa436fb45313e6d36c4302 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2c6798e56d5490ab4e0b37a16ab2590 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c6a6309b14a4beab788166a9dbd199c : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7861b638dc524cb19f8dac5f62e5d220 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3f2d731f8b0ddcb9015980c60a30cacc67840b8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 04:01:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"83452d350306d85c956efdbb01b75adbca7b45e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 04:02:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb723e1251e6403aa8908ac06303ee7f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01ab65208b56ab00a0ab8c915052bfdab6d7e0a4","revisions":{"01ab65208b56ab00a0ab8c915052bfdab6d7e0a4":{"kind":"REWORK","_number":1,"created":"2020-06-08 23:23:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/73/734373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/73/734373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/73/734373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/73/734373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/73/734373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/73/734373/1"}}},"commit":{"parents":[{"commit":"d137023727cfd117f813f96aaf0f7cb235d29c5a","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/d137023727cfd117f813f96aaf0f7cb235d29c5a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:23:17.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:23:17.000000000","tz":120},"subject":"drop mock from lower-constraints","message":"drop mock from lower-constraints\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we don\u0027t need it\nin lower-constraints.\n\nThese changes will help us to drop `mock` from openstack/requirements\n\nChange-Id: I7f0cc935291e5c0c4f905ad2aa0fb03e6192abb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/01ab65208b56ab00a0ab8c915052bfdab6d7e0a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/01ab65208b56ab00a0ab8c915052bfdab6d7e0a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
