)]}'
{"id":"openstack%2Fmistral-lib~895468","triplet_id":"openstack%2Fmistral-lib~stable%2F2023.2~I17d26af1dcedf0f11093b09ebd4eec8f8326530d","project":"openstack/mistral-lib","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"I17d26af1dcedf0f11093b09ebd4eec8f8326530d","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","status":"MERGED","created":"2023-09-15 14:44:24.000000000","updated":"2023-10-06 08:29:28.000000000","submitted":"2023-10-06 08:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895468-create-2023.2","meta_rev_id":"eca4bd9a96198ad868d74e85238e7fe56a74a8d8","_number":895468,"virtual_id_number":895468,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 08:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":2,"date":"2023-09-18 14:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2023-10-06 07:48:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2023-09-21 16:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-09-19 14:40:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2023-10-06 07:48:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 15:06:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-18 14:04:48.000000000","updated_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2023-09-19 14:40:28.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"state":"REVIEWER"},{"updated":"2023-09-21 16:19:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-10-06 07:48:58.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"bd0c76badd58e6954aa1338d73e9168c6502abe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-15 14:44:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58bcaa5cf39947427e5e6e5b8e7391a357187382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 15:06:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d3f0017a4148c19f8314b17c45e6da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711ce6d414e4489a8e9c428880767374 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcc3197a86484ddc90b05c634b02c800 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/099f80e7cd2b47de83066c677cf15c96 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae288887af443639ad30b45b3b3aec6 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c72de92a452b413db06738afd3e3668b : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a58dcf90227144b28f63b8c60a15d77e : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4312cb68d2a4928aac85e7403ce3dd6 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"94e10cf380dcc1d2fdb895e59a6e2eeff1094e20","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2023-09-18 14:04:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d9bacbb33925f019f3531754b957ce84ae8e7a59","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-19 14:40:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"71172d4e32d94953ca101bf515909bb230dcd933","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-21 16:19:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf8e1d26b424b6b529da7ac18b2b7ed698b69948","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2023-10-06 07:48:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93e1fb9f4832ceb52f5115152ee7db4233d7a593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 07:49:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fb337633f268b9aadc0d8bb7bbb756f274c6664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 08:27:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e9edbae38504151bfd50a907e45d8c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d99edfd5244df6a7a23674b4fb8f90 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7956fd69aba4a98ad592cf8cb5764bf : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2caa7dafd234b7ea62c9bbeb5f0a18c : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e0755e0c2f44c8498d2396e68ae7789 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f3a95c59254b24b2ece549bcd60e23 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6637d6959f934bfb915a2f7a9e132a4a : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e7cd1c0402318e3a96d25951e7d2f1cae513df31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 08:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eca4bd9a96198ad868d74e85238e7fe56a74a8d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 08:29:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a7b87b025b4a3f91bcec8ed775bebe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0739e8f90ea7431194e59a255a8e3765 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/416093e780434c74b83cb57b37c2c20b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf628267fdcfe8c55b0d600c2b1a045a87f73a17","revisions":{"bf628267fdcfe8c55b0d600c2b1a045a87f73a17":{"kind":"REWORK","_number":1,"created":"2023-09-15 14:44:24.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/68/895468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/68/895468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/68/895468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/68/895468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/68/895468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/68/895468/1"}}},"commit":{"parents":[{"commit":"e7e939d5692a93315e3b59a2b3a924a61a8a0a10","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/e7e939d5692a93315e3b59a2b3a924a61a8a0a10"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:44:21.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:44:21.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I17d26af1dcedf0f11093b09ebd4eec8f8326530d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/bf628267fdcfe8c55b0d600c2b1a045a87f73a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/bf628267fdcfe8c55b0d600c2b1a045a87f73a17"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
