)]}'
{"id":"openstack%2Fmistral-lib~914371","triplet_id":"openstack%2Fmistral-lib~master~I89634eecf93fbfe9ebcd0cf06235a27e1d3878dd","project":"openstack/mistral-lib","branch":"master","topic":"reno-eom-victoria","hashtags":[],"change_id":"I89634eecf93fbfe9ebcd0cf06235a27e1d3878dd","subject":"reno: Update master for unmaintained/victoria","status":"MERGED","created":"2024-03-26 17:51:07.000000000","updated":"2024-04-08 06:54:53.000000000","submitted":"2024-04-08 06:53:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914371-reno-eom-victoria","meta_rev_id":"336af1bf3d47a3fe014c015f9f30d2a27e148927","_number":914371,"virtual_id_number":914371,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-08 06:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2024-04-08 06:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2024-04-08 06:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 18:00:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-08 06:45:55.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"f2c5c1744e15d605ec95ea9e478ec28c87a564be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-26 17:51:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11c9bd510aca5ead06fe906ba1ab400f93aeaa6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 18:00:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/176bfb5848ce48cd8b0ac7833efee5f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7276d7cd18345329685babb8f2edd67 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c46ba12c994717a015670b4830a4ae : SUCCESS in 2m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4e5c9e97eba4168ad7af0c71b1b86b2 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"39a426696afa15aedf82f5caffb60a7531c4009f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2024-04-08 06:45:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41d99e95d37ce965edc64c5edf3f95dc8fc0ac9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:46:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"098e419da28c82b5328d37a59749ee8549ba41e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:53:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d453e52afb6646dd87ce89d297cf7c93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c1d8ccd0cb480192806639c09fb7ae : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800c5faae15e4a4fbab2db098ead68bc : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb95e22c36e14f398e94b19e14150bcd : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1b349958dd1c4efbe1b6a0abe4689e0728af3a12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:53:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"336af1bf3d47a3fe014c015f9f30d2a27e148927","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:54:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fdcb2c18f5547889c71f4cfdf553ff3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6b1e06837441c68503678744c5cec1 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e96904c5f54941b7fbe56ab507d3c4 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68bd847a71d9f3598faa499a58379785b34de1ef","revisions":{"68bd847a71d9f3598faa499a58379785b34de1ef":{"kind":"REWORK","_number":1,"created":"2024-03-26 17:51:07.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/71/914371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/71/914371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/71/914371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/71/914371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/71/914371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/71/914371/1"}}},"commit":{"parents":[{"commit":"b4c15e4e9e279514e5f2c7689c7a049f9ff35539","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/b4c15e4e9e279514e5f2c7689c7a049f9ff35539"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-26 17:51:05.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-26 17:51:05.000000000","tz":0},"subject":"reno: Update master for unmaintained/victoria","message":"reno: Update master for unmaintained/victoria\n\nUpdate the victoria release notes configuration to build from\nunmaintained/victoria.\n\nChange-Id: I89634eecf93fbfe9ebcd0cf06235a27e1d3878dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/68bd847a71d9f3598faa499a58379785b34de1ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/68bd847a71d9f3598faa499a58379785b34de1ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
