)]}'
{"id":"openstack%2Fmistral-lib~914382","triplet_id":"openstack%2Fmistral-lib~master~I8f919410e677cd2ce53ca242bb9b7a5690340c89","project":"openstack/mistral-lib","branch":"master","topic":"reno-eom-wallaby","hashtags":[],"change_id":"I8f919410e677cd2ce53ca242bb9b7a5690340c89","subject":"reno: Update master for unmaintained/wallaby","status":"MERGED","created":"2024-03-26 17:52:37.000000000","updated":"2024-04-08 06:54:08.000000000","submitted":"2024-04-08 06:53:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914382-reno-eom-wallaby","meta_rev_id":"2c0f341bf4e90ad6729299c03b6b58e48aec16ee","_number":914382,"virtual_id_number":914382,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-08 06:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2024-04-08 06:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2024-04-08 06:45:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 18:09:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-08 06:45:00.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"}],"messages":[{"id":"896e2544a7e61ca451c3cdcb46830d1f37cf8922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-26 17:52:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9740a7fa5b51b41b2a5579fc0a59a132a064d68b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 18:09:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85960e99976a46b8baee681cbf0baebd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0be7695a9a4785a88e3d628dfd4bae : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c42591a13940c1b2987389b127c594 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7831dfae7aa345c6a3694a1a67deb8d8 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"214de8e418e51a4501b0d8e48649611559c8bb67","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2024-04-08 06:45:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20572921f1af4145ba6334a937a50b71d30da18d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:45:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e5e1d16557442b95635ac8c856f87d6d736025b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:53:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26c6742080784c9e80f21fc1202fd5ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/929ca4277dde4b1cae133ae6563674cd : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a4ad450827482eb9663a272b1fe771 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e9399baafce4f8ca87afb97b6075839 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1},{"id":"891d66f9927987a4cb931a83401dcc79b8fa0fa4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:53:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c0f341bf4e90ad6729299c03b6b58e48aec16ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:54:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b674a78f5a4ea1bb6b4509fd504ebc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc295d796b14b419a05bc699dd033ec : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69f4735112434dfa997ab2a88b68a06d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab500b73afe270f9f625460d7db2e4cdd02002b3","revisions":{"ab500b73afe270f9f625460d7db2e4cdd02002b3":{"kind":"REWORK","_number":1,"created":"2024-03-26 17:52:37.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/82/914382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/82/914382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/82/914382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/82/914382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/82/914382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/82/914382/1"}}},"commit":{"parents":[{"commit":"b4c15e4e9e279514e5f2c7689c7a049f9ff35539","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/b4c15e4e9e279514e5f2c7689c7a049f9ff35539"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-26 17:52:35.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-26 17:52:35.000000000","tz":0},"subject":"reno: Update master for unmaintained/wallaby","message":"reno: Update master for unmaintained/wallaby\n\nUpdate the wallaby release notes configuration to build from\nunmaintained/wallaby.\n\nChange-Id: I8f919410e677cd2ce53ca242bb9b7a5690340c89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/ab500b73afe270f9f625460d7db2e4cdd02002b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/ab500b73afe270f9f625460d7db2e4cdd02002b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
