)]}'
{"id":"openstack%2Fmistral-lib~943724","triplet_id":"openstack%2Fmistral-lib~stable%2F2025.1~I85efa67e567931559be441b6e096035af0c9a1d3","project":"openstack/mistral-lib","branch":"stable/2025.1","topic":"create-2025.1","hashtags":[],"change_id":"I85efa67e567931559be441b6e096035af0c9a1d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","status":"MERGED","created":"2025-03-07 14:29:33.000000000","updated":"2025-03-08 18:42:50.000000000","submitted":"2025-03-08 18:40:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943724-create-2025.1","meta_rev_id":"0b48af7beb8b7f22404b64b8470b733e5703ede0","_number":943724,"virtual_id_number":943724,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-08 18:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-08 18:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-08 18:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 15:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-08 18:18:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"145d3da9ae04e1deabab1388e776fe6b1739ac44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-03-07 14:29:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc57b13acdb5244e9fb8ff679a20a6aad5b9953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 15:05:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33b03fff2974df7ba79ee77c76459bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09f538d2cade4bf0bae221570299d9e0 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4d3e0a1f32a49249bee21522c77d7b9 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6c72f2d8ffd440a831030aca21ae257 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9b2801cd614b949734e365a042bbfb : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3c0b1da6b484d6291a0d13fc06368a7 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cccd31d86455aff69c3e660d92d06f9d4c6cfccb","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-08 18:18:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3051683414784c186023f876666f7fadb7e19d33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 18:18:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71ea58d28f164c023e0eb80a9d82aab414b711c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 18:40:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7734a532efa42f99a57e4b96abb576b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b789e07c32e5418b861986c6ac360ba3 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d02487a193234d3ba2bd02cec39d071f : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30e56fc52408492c9a1e5a5b814f5329 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b19219cbd54d22a2190d1ee89c3755 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce6fdc422bea45f58e186e746c7974f2 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0c7dae919fe2f3f5308726b05704a4cd13c06af9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 18:40:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0b48af7beb8b7f22404b64b8470b733e5703ede0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 18:42:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941a82a67125442bb4432fddd8962b98\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e513e7cba4490e993e14415cf24be0 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11943b79956549c2a76d4d2fede82c36 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"693174dd0aac1da22870b31e4a2481c4e749916a","revisions":{"693174dd0aac1da22870b31e4a2481c4e749916a":{"kind":"REWORK","_number":1,"created":"2025-03-07 14:29:33.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/24/943724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/24/943724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/24/943724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/24/943724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/24/943724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/24/943724/1"}}},"commit":{"parents":[{"commit":"c3d648d4a91ef7afb4ad4f585d7b53eb5aaf867d","subject":"Update .gitreview for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/c3d648d4a91ef7afb4ad4f585d7b53eb5aaf867d"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-07 14:29:32.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-07 14:29:32.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2025.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2025.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I85efa67e567931559be441b6e096035af0c9a1d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/693174dd0aac1da22870b31e4a2481c4e749916a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/693174dd0aac1da22870b31e4a2481c4e749916a"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
