)]}'
{"id":"openstack%2Fmistral-lib~952030","triplet_id":"openstack%2Fmistral-lib~master~Ic29bd97b6c09c5f4511ea6d3f410ea242bdd8cbe","project":"openstack/mistral-lib","branch":"master","topic":"license-classifier","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-06-09 07:58:19.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-06-09 07:35:09.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ic29bd97b6c09c5f4511ea6d3f410ea242bdd8cbe","subject":"Replace license classifier","status":"MERGED","created":"2025-06-08 13:26:40.000000000","updated":"2025-06-09 07:59:03.000000000","submitted":"2025-06-09 07:58:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952030-license-classifier","meta_rev_id":"570552c8e3fdb9ebeaf60cbd7b6d6ab741d2d462","_number":952030,"virtual_id_number":952030,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 07:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-09 07:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-09 07:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-08 13:27:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-06-08 13:36:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78fe10ede0c99a622c0603a69313e5a9dfa81c90","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:26:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04eb6d43521d7bce7a8659576b454fff624d9fcc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:27:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a73ac3619299427b04b221c908e834db4f9e03b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:27:35.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c1ef9809e0f3c37124d650a4be6ef913a5724c82","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-08 13:27:41.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"034414dfef67de3d16af1e57456064ffa0d0164a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 13:36:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07d8a8555f6a436485a2f4988bd572ca\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5831a7c70144cdb818abc9466cb2a46 : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353a7af99fc14fe9b8f402bef52d1cde : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39c14ce70287403a9e4b7ffca9980740 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1d799e5a4ce4c52ac272c04414551b4 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9abac583e8574bf0b8b60a52ae17b6de : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f215cabb03754d0090d7c3a1d7687585 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":3},{"id":"ed85b5ecd6321f8f49aebf348bdc767fa95dad82","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-06-09 07:35:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bf483736129742e7a68462f8286fb2487e666eb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:35:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"834ed353ec5efb54c196b3bc84d84c0dd6a4f867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:58:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/208f2f21e4ce4e0cbd9fdc0d5262bce6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9165cc4c1d140e6b8c511c2c98ece84 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c19ac810f0460881854e014a1d7aed : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d43302ed2bb84bd184fd2a93d64d2b6f : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aaa13dcce98b42a8b9269cc252801e54 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9686d7dcc6647f39fa39fa067a90d15 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"575a1d6bebcb6f663daae5802446e8f0a0193f34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:58:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"570552c8e3fdb9ebeaf60cbd7b6d6ab741d2d462","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 07:59:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a15e87df53543afa1bf245d6b9dbb14\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6da8fc30f5456894ab9b5df53903c1 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5422adaaa79b7f1760b3efd6a37787d0f8cb8d63","revisions":{"39ebf654ad9f61f50a5fd4fe2ed08b82488e2174":{"kind":"REWORK","_number":1,"created":"2025-06-08 13:26:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/30/952030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/30/952030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/1"}}},"commit":{"parents":[{"commit":"7372311e1893a6cc466b213ba19dcc7a3e015b84","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/7372311e1893a6cc466b213ba19dcc7a3e015b84"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:26:40.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:26:40.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic29bd97b6c09c5f4511ea6d3f410ea242bdd8cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/39ebf654ad9f61f50a5fd4fe2ed08b82488e2174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/39ebf654ad9f61f50a5fd4fe2ed08b82488e2174"}]},"branch":"refs/heads/master"},"2d8af5f3a4f691a6e3cc6dab368ca204a946d1b3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-08 13:27:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/30/952030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/30/952030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/2"}}},"commit":{"parents":[{"commit":"7372311e1893a6cc466b213ba19dcc7a3e015b84","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/7372311e1893a6cc466b213ba19dcc7a3e015b84"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:26:40.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:27:06.000000000","tz":0},"subject":"Replace license classifier","message":"Replace license classifier\n\n... according to the following warning from setuptools.\n\nSetuptoolsDeprecationWarning: License classifiers are deprecated.\n\n********************************************************************************\nPlease consider removing the following classifiers in favor of a SPDX license expression:\n\nLicense :: OSI Approved :: Apache Software License\n\nSee https://packaging.python.org/en/latest/guides/writing-pyproject-toml/#license for details.\n********************************************************************************\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: Ic29bd97b6c09c5f4511ea6d3f410ea242bdd8cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/2d8af5f3a4f691a6e3cc6dab368ca204a946d1b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/2d8af5f3a4f691a6e3cc6dab368ca204a946d1b3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5422adaaa79b7f1760b3efd6a37787d0f8cb8d63":{"kind":"REWORK","_number":3,"created":"2025-06-08 13:27:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/30/952030/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/30/952030/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/30/952030/3"}}},"commit":{"parents":[{"commit":"7372311e1893a6cc466b213ba19dcc7a3e015b84","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/7372311e1893a6cc466b213ba19dcc7a3e015b84"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:26:40.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-06-08 13:27:35.000000000","tz":0},"subject":"Replace license classifier","message":"Replace license classifier\n\n... according to the following warning from setuptools.\n\nSetuptoolsDeprecationWarning: License classifiers are deprecated.\n\n********************************************************************************\nPlease consider removing the following classifiers in favor of a SPDX license expression:\n\nLicense :: OSI Approved :: Apache Software License\n\nSee https://packaging.python.org/en/latest/guides/writing-pyproject-toml/#license for details.\n********************************************************************************\n\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: Ic29bd97b6c09c5f4511ea6d3f410ea242bdd8cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/5422adaaa79b7f1760b3efd6a37787d0f8cb8d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/5422adaaa79b7f1760b3efd6a37787d0f8cb8d63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
