)]}'
{"id":"openstack%2Fmistral-lib~957369","triplet_id":"openstack%2Fmistral-lib~master~Idddfaf500d0eaddfcaad1b71f4b31fc7b910df9b","project":"openstack/mistral-lib","branch":"master","topic":"release-health-check-gazpacho","hashtags":[],"change_id":"Idddfaf500d0eaddfcaad1b71f4b31fc7b910df9b","subject":"DNM: gate health test","status":"ABANDONED","created":"2025-08-14 09:43:43.000000000","updated":"2026-02-23 13:48:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee7bbaba6d59a057a72ba4d11e3f20c2d73a5b20","_number":957369,"virtual_id_number":957369,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-13 11:36:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 23:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5777569cea48932522231480fd6c0a20f9886a27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 09:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de36e53e25815a35ff1c0a6cabb8e234af6d53b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 23:35:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12962bc1fd93400ab4b21c6a193fce15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6663e11fff4f219cf9adb96f5636b9 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d8cff2023884e69b78f5ce2ed547ed1 : SUCCESS in 2m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b4d99aba25740568aaf6f4b3e8f880d : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c7247c91ab549cc85bc29002defe91d : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6310995c731e4df4a3acc5524964c8bc : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"170d97ce2ece6255743ffeb414ab110beae5abad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-15 07:52:53.000000000","message":"Abandoned\n\ngate jobs are passing, patch can be abandoned.","accounts_in_message":[],"_revision_number":1},{"id":"302bf35309838f3129e05624997924920c6ee1b6","tag":"autogenerated:gerrit:restore","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-13 11:25:51.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"078f12a3acde1357b7a367cf0c0b452184e365d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-13 11:26:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1e01bff403ca5d78b143ae18d7a43ac2dd619357","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-13 11:26:11.000000000","message":"Topic release-health-check-flamingo removed","accounts_in_message":[],"_revision_number":2},{"id":"7bbbc4c9deff8212e63c98dd1fc8de0b499262b2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-13 11:26:19.000000000","message":"Topic set to release-health-check-gazpacho","accounts_in_message":[],"_revision_number":2},{"id":"7d77f084e2ae299853519343d60a0e7919fabfeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 11:36:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcf64e373d25428fbbefa88dfe665dd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e745470bf4417abca0c44d84763887 : SUCCESS in 1m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/728f90af88794b61a66854fea52d2629 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d4b9b9933b246f8b51e32faddc24e6a : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dcaecad27fb4428a20a147fda58c126 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ee7bbaba6d59a057a72ba4d11e3f20c2d73a5b20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-23 13:48:11.000000000","message":"Abandoned\n\ngate jobs passed. abandoning.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4fdd7cbe75b9e006fedccad87e612acdc9d2225d","revisions":{"a569a6bc022739f511930853b1f70d065575eb10":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:43:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/957369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/69/957369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/1"}}},"commit":{"parents":[{"commit":"c98e93fc6dcdbb7fbd1ed6756054c2561ca394c1","subject":"Merge \"Drop support for Python 3.8 and 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/c98e93fc6dcdbb7fbd1ed6756054c2561ca394c1"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:43:34.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:43:42.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Idddfaf500d0eaddfcaad1b71f4b31fc7b910df9b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/a569a6bc022739f511930853b1f70d065575eb10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/a569a6bc022739f511930853b1f70d065575eb10"}]},"branch":"refs/heads/master"},"4fdd7cbe75b9e006fedccad87e612acdc9d2225d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-13 11:26:06.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/69/957369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-lib","ref":"refs/changes/69/957369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-lib refs/changes/69/957369/2"}}},"commit":{"parents":[{"commit":"6db58e38adbe6b34bc48fb24c0c333d8ccc35dad","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/6db58e38adbe6b34bc48fb24c0c333d8ccc35dad"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:43:34.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-02-13 11:26:06.000000000","tz":0},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: Idddfaf500d0eaddfcaad1b71f4b31fc7b910df9b\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/4fdd7cbe75b9e006fedccad87e612acdc9d2225d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-lib/commit/4fdd7cbe75b9e006fedccad87e612acdc9d2225d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
