)]}'
{"id":"openstack%2Fmistral-tempest-plugin~1005312","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I6645fc02a1d4ec845afb208d6636dc764b2a29a2","project":"openstack/mistral-tempest-plugin","branch":"master","topic":"2026.2-job","hashtags":[],"change_id":"I6645fc02a1d4ec845afb208d6636dc764b2a29a2","subject":"Add stable/2026.2 job","status":"NEW","created":"2026-09-11 17:09:31.000000000","updated":"2026-09-11 18:04:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"15d215b25e1aa0e0ae89e9cae0f4c562c1dafac3","_number":1005312,"virtual_id_number":1005312,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 18:04:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 18:04:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bcedf3cb1220b228ed7f5341f0f694e13a80066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-11 17:09:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15d215b25e1aa0e0ae89e9cae0f4c562c1dafac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:04:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8624ce0700d3439abf7c83bc39883c04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f16eba086944803b77dc745c8f72baf : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9bfc6b643824937946fc2e2c0f7236f : SUCCESS in 2m 32s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8f9c6b14bace46d89be17b4fa3250d53 : SUCCESS in 21m 16s\n- mistral-devstack-2026-2 https://zuul.opendev.org/t/openstack/build/53393acafb524dcab20f63cf253c2685 : SUCCESS in 19m 13s\n- mistral-devstack-2026-1 https://zuul.opendev.org/t/openstack/build/a4d7b426ba01448a85f5e87f3b8bd0ae : SUCCESS in 27m 07s\n- mistral-devstack-2025-2 https://zuul.opendev.org/t/openstack/build/71a386c8630c41f7bdefc20d626f58d4 : SUCCESS in 47m 57s\n- mistral-devstack-2025-1 https://zuul.opendev.org/t/openstack/build/8cefb65728ec408ba31ca4d70eba4b21 : SUCCESS in 20m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b5f29dd26ae461d8253b0bad5d2cb4c : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c76491d1b844a93ef98c81ad5e02d56769b3805","revisions":{"9c76491d1b844a93ef98c81ad5e02d56769b3805":{"kind":"REWORK","_number":1,"created":"2026-09-11 17:09:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/1005312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/12/1005312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/12/1005312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/12/1005312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/12/1005312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/12/1005312/1"}}},"commit":{"parents":[{"commit":"1115f1e870a446bb7e929da19b41be08a20d86ca","subject":"Merge \"Restore explicit package list for module detection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/1115f1e870a446bb7e929da19b41be08a20d86ca"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:09:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-11 17:09:20.000000000","tz":540},"subject":"Add stable/2026.2 job","message":"Add stable/2026.2 job\n\nThe branch was created so we should start testing it.\n\nChange-Id: I6645fc02a1d4ec845afb208d6636dc764b2a29a2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/9c76491d1b844a93ef98c81ad5e02d56769b3805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/9c76491d1b844a93ef98c81ad5e02d56769b3805"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
