)]}'
{"id":"openstack%2Fmistral-tempest-plugin~678405","triplet_id":"openstack%2Fmistral-tempest-plugin~master~Idc39d6f6e367d3dea673d48a160b8314a51394e8","project":"openstack/mistral-tempest-plugin","branch":"master","topic":"eyalb/py3","hashtags":[],"change_id":"Idc39d6f6e367d3dea673d48a160b8314a51394e8","subject":"use python3 for gate","status":"MERGED","created":"2019-08-25 08:10:58.000000000","updated":"2019-08-28 06:49:18.000000000","submitted":"2019-08-28 06:49:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678405-1566974958094-727564a4","meta_rev_id":"020b3b822c1154ea561a274923ebb6abf42b0326","_number":678405,"virtual_id_number":678405,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-08-28 06:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2019-08-25 12:46:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2019-08-28 05:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-08-28 05:48:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-25 12:46:24.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-08-28 05:48:46.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-08-28 06:49:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4275a12e1269b69943ead2922cd80e577b3ecebd","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-25 08:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8bf782a936adfaf45802de865af93dfa6bb0e2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 08:56:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab706fb7d0d447a4a3935fe940f795f5 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6722e4d6b4a945e39f0ae761fc368787 : SUCCESS in 3m 52s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/04be665300414066970394d77e9c0967 : FAILURE in 42m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b369e00c71b146419bbb34900996f930 : FAILURE in 40m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/107caea26986449f8389be8017bd39a1 : FAILURE in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"12494bee62a19142d19231a59c03a51250d0ed4f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-25 09:57:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ccd3fc3e79e2bde1dbce9f71bd02215ec367340","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 10:46:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2efbe9759f67433897a4f4a58fdf23db : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff06c62eb38403f923dce28f7975972 : SUCCESS in 3m 18s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/59465aca24f6481fad7da0e927467ea6 : SUCCESS in 43m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/697a0b2a5660491c92fa2ba54b2f7d65 : SUCCESS in 41m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ed61b22718a545c99ba36e2fc3a82574 : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":2},{"id":"bcc4b7baaa6cc87262f27be2a7e038918d9105bb","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-08-25 12:46:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3deec58558780889b58821ac45ec5c3502cde211","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-08-28 05:48:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b57fa0ae761b88adcfb7fab6066888574872f6d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 05:49:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc1cce5f163d0a03b70edd4ccae2bd49f7551b1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 06:49:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38d9b3fbeb2497ca18e8271d944f4ca : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb750f5db114a749cc74958b0b41604 : SUCCESS in 4m 19s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e3b81b9933f34bdc8202f52fff619036 : SUCCESS in 42m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4852b1df8c204d18ab7c28eaef604a3c : SUCCESS in 44m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/06326a10f35c41e9b65d070c28120302 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":2},{"id":"eb8991c8a9bc2b31ea2890d7475cedde0f354517","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 06:49:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"87112692424b98af5b18701dff0b0ac90d2c215c","revisions":{"81af20a4af51fceffac95f4675a0cdb52ca09fab":{"kind":"REWORK","_number":1,"created":"2019-08-25 08:10:58.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/05/678405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/05/678405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/1"}}},"commit":{"parents":[{"commit":"dee6cd7570442c19050b0cf2939081fde4ca287a","subject":"Fix mistral-devstack quoting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/dee6cd7570442c19050b0cf2939081fde4ca287a"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 08:09:32.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 08:10:49.000000000","tz":180},"subject":"use python3 for gate","message":"use python3 for gate\n\nChange-Id: Idc39d6f6e367d3dea673d48a160b8314a51394e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/81af20a4af51fceffac95f4675a0cdb52ca09fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/81af20a4af51fceffac95f4675a0cdb52ca09fab"}]},"branch":"refs/heads/master"},"87112692424b98af5b18701dff0b0ac90d2c215c":{"kind":"REWORK","_number":2,"created":"2019-08-25 09:57:24.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/05/678405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/05/678405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/05/678405/2"}}},"commit":{"parents":[{"commit":"dee6cd7570442c19050b0cf2939081fde4ca287a","subject":"Fix mistral-devstack quoting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/dee6cd7570442c19050b0cf2939081fde4ca287a"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 08:09:32.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-08-25 09:56:27.000000000","tz":180},"subject":"use python3 for gate","message":"use python3 for gate\n\nChange-Id: Idc39d6f6e367d3dea673d48a160b8314a51394e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/87112692424b98af5b18701dff0b0ac90d2c215c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/87112692424b98af5b18701dff0b0ac90d2c215c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
