)]}'
{"id":"openstack%2Fmistral-tempest-plugin~716898","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I19b57e686306b9d3f26bc5a0228ebf5d7c08bb20","project":"openstack/mistral-tempest-plugin","branch":"master","hashtags":[],"change_id":"I19b57e686306b9d3f26bc5a0228ebf5d7c08bb20","subject":"Setting Python version explicitly to 3 for devstack jobs","status":"MERGED","created":"2020-04-02 08:49:48.000000000","updated":"2020-04-02 16:19:55.000000000","submitted":"2020-04-02 16:19:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716898-1585844395871-38ce0ed0","meta_rev_id":"63121d042af67e1d3883f955775e915d99a39356","_number":716898,"virtual_id_number":716898,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 16:19:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-04-02 14:57:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-02 12:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-04-02 15:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 12:39:01.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-02 14:57:57.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-04-02 16:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c66ffc4851e4ea21aaa938d15e596cbea614447c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 08:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c0ac074400c56a65319d407a62029a442004820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 09:48:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f7ca2e97584263a9abb599e22ab1da : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5065ca0a7a1a4d368389ef912834447e : SUCCESS in 4m 55s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8ec4579c6342428985dec8dcce7cf3cc : SUCCESS in 43m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c6ee7dbd30b4538ba3b953db93b9ea7 : SUCCESS in 46m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df4c928f3b1d4aa5951540782988a7b0 : SUCCESS in 44m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d958f7fbbff9409eb961f649ef46be6b : SUCCESS in 48m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4af13b1114fa4bce9f38daa28d74a82c : FAILURE in 55m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7b6334cb7f8f7728dc6d87b94fe506038c6a74b1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 10:55:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e592bb7e5278b14d26d4f4b40d71cd3e82373aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:56:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57eb9eee6e1d4a70b71dc8ea6423098b : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03433d811ce0400ea145be0101f1f449 : SUCCESS in 4m 57s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fd5c8126d0764b97878ebcf401a07e53 : SUCCESS in 47m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a4ba97efe9438e88082bd96128b9eb : SUCCESS in 45m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d8c95d0081d447486c7083362c387ce : SUCCESS in 48m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7a2428b672bd405a83a474c644df9517 : SUCCESS in 45m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ddbf29ef00e34355b3d0a556e39e6444 : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d8205e4a97275b6affd5c9515ef09a708a9a8884","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-02 12:39:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23052a85c908755a0bfef339b663824c535b7d64","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-04-02 14:57:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d32cb78bf26c03f18dba30f7b69667e999aedab3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 15:02:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28a49f0c5cefa67f27bdc8a130dd22bd534e2633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 15:03:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d48438a919ecd9634d75d922538b22c3dc503bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:19:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea9275499c5b43c3b29284563e22f417 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c82a21ee40b4e53bbce4ce833421c66 : SUCCESS in 6m 36s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b7627131fd284c61b9f33cb56abd05a6 : SUCCESS in 51m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5ff6420e35e49b7846a73323c68e740 : SUCCESS in 49m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef8b39b21b9f4179a8dc930765bc9f8e : SUCCESS in 52m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/14c91b36a6a4486aaef501f1b21778e1 : SUCCESS in 52m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cf039e03100f42aab6e958f04885a83a : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":1},{"id":"63121d042af67e1d3883f955775e915d99a39356","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:19:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ff9a7efc4d1127634389eb0e5bdabe3ec63570b","revisions":{"3ff9a7efc4d1127634389eb0e5bdabe3ec63570b":{"kind":"REWORK","_number":1,"created":"2020-04-02 08:49:48.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/98/716898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/98/716898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/716898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/716898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/716898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/716898/1"}}},"commit":{"parents":[{"commit":"6f746b92282f1a4b82509448d53130a74adeb0c4","subject":"Fix preinstalled workflows tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/6f746b92282f1a4b82509448d53130a74adeb0c4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-02 08:47:53.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-02 08:47:53.000000000","tz":420},"subject":"Setting Python version explicitly to 3 for devstack jobs","message":"Setting Python version explicitly to 3 for devstack jobs\n\n* Backports to stable/train are now failing because tempest requires\n  Py3 but for some reason Py2 is provided. So we have to explicitly\n  define the version of Python in master because there are no\n  stable branches for tempest.\n\nChange-Id: I19b57e686306b9d3f26bc5a0228ebf5d7c08bb20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/3ff9a7efc4d1127634389eb0e5bdabe3ec63570b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/3ff9a7efc4d1127634389eb0e5bdabe3ec63570b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
