)]}'
{"id":"openstack%2Fmistral-tempest-plugin~725693","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I9498979712375715eab79e77a1c0e89d0efe33b5","project":"openstack/mistral-tempest-plugin","branch":"master","topic":"stable-testing-tempest-plugins","hashtags":[],"change_id":"I9498979712375715eab79e77a1c0e89d0efe33b5","subject":"Add stable branch jobs on the plugins master gate","status":"MERGED","created":"2020-05-05 22:16:05.000000000","updated":"2020-05-12 07:09:46.000000000","submitted":"2020-05-12 07:09:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725693-1589267386321-793a8d12","meta_rev_id":"6a499b520e646dc5bcd737d5c6a0cf986bfbb4fc","_number":725693,"virtual_id_number":725693,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"date":"2020-05-06 00:28:00.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 07:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-05-12 05:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-11 11:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-05-12 05:49:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 11:32:32.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-05-12 05:49:47.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-05-12 07:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfdcddbce7207ba5f6275b52638249f6c0cc730a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 22:16:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2bea62799ae1adca2bc82e79a0818ee4d6ca28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 23:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/089c659cd97e4d6fae7aef138e09d66d : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63c892dcb604f76834e712e73e5d476 : SUCCESS in 3m 47s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d332f7e9173749918e907164ded7ea15 : FAILURE in 36m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2986356119b4e35ba4f3455c336395f : FAILURE in 39m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92cd894253aa4dd2a6123e92d64178c1 : FAILURE in 37m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a6325f1ccbaa4b19b70246c8fe876623 : FAILURE in 35m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/042388f426fd4b13bd5de1d5cf8e9248 : FAILURE in 35m 27s\n- mistral-devstack-ussuri https://zuul.opendev.org/t/openstack/build/35a290c347b442f8b1e207d777bc69dc : SUCCESS in 43m 20s\n- mistral-devstack-train https://zuul.opendev.org/t/openstack/build/2272c421fb114b82b4084561b83f7df0 : SUCCESS in 47m 28s\n- mistral-devstack-stein https://zuul.opendev.org/t/openstack/build/8f28b3d8faac462f88a03c19a2a4d2e2 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a9fd5b05406c9d758a06dd63a5c1d70b08e1dfbf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-06 00:28:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc8a6c9521384739dac55c98f39b2668f1c5f69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 01:18:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f33d9feb2f34c5b9688a35672a12ed2 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8200780749a14c598c98b9cc2809efd0 : SUCCESS in 3m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/35d159794aad4e5690357207aaaa5733 : FAILURE in 42m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3630ed86a83488d982dd5cab09cbaf8 : FAILURE in 43m 13s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5fbbacfb1fb49e786458fd4a8081634 : FAILURE in 35m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/af2ac9b35150487f8666eb3f199222bd : FAILURE in 42m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/92f651bd5326444e834d01ccc6649d8c : FAILURE in 41m 19s\n- mistral-devstack-ussuri https://zuul.opendev.org/t/openstack/build/3f7a52bd3fc541978d5bd134034dfb94 : SUCCESS in 48m 07s\n- mistral-devstack-train https://zuul.opendev.org/t/openstack/build/a13e820c78e54385b8e1c602f1e50e70 : SUCCESS in 38m 05s\n- mistral-devstack-stein https://zuul.opendev.org/t/openstack/build/cedefa15a38a42f18fa9ffaf182b0a69 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":1},{"id":"50c8d38a93251c317ffe719cb1ef3c88bcb11de2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-09 13:25:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76620ac0d1b31df61e22731e629ee8dbcecc3855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 14:11:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392f43d504a84fad92ea5f5c345a4575 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ee7fec0f554b209fbe648f51b437d3 : SUCCESS in 4m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9b0b68d8a68d4c3caf62ae8018d56c8f : FAILURE in 37m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6af564a27d84b85927fe4ac69e45666 : SUCCESS in 39m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d4d220b18b4cd7b7297d9e537a9cf2 : SUCCESS in 43m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c78fe618f83f4e169d6044faf9d648e8 : SUCCESS in 38m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6db21667ffeb43a4856e5d4028c726ef : SUCCESS in 39m 32s\n- mistral-devstack-ussuri https://zuul.opendev.org/t/openstack/build/b995926896a541ada6d955ee4b01dce5 : SUCCESS in 38m 32s\n- mistral-devstack-train https://zuul.opendev.org/t/openstack/build/1388eb11887d41d99e56c7c686cd297d : SUCCESS in 41m 12s\n- mistral-devstack-stein https://zuul.opendev.org/t/openstack/build/cd9cc226d04541b09239ca1831b9ec2e : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9284d8f8dc0b70971a5137a2e541ebe071bca95f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-09 14:19:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67300a40a0cd2d8c1d11e91eece9308b5b84cc25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 15:14:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aca156c6e1844068b0eee0480bce2fc : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff764c6a17ce47a08ed8c1ecb94cc8a3 : SUCCESS in 3m 08s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ee31e40b18be491c973adfc068c78ff6 : SUCCESS in 36m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c06394a94d9a49038ea0b80471b1d96f : SUCCESS in 35m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5eb498ad4c646ad9f50d7590fcc2e14 : SUCCESS in 41m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2c91ad2dc2c44c9fbe9f4d031e739f35 : SUCCESS in 54m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/427da087718e4ac58fac834533429881 : SUCCESS in 54m 08s\n- mistral-devstack-ussuri https://zuul.opendev.org/t/openstack/build/5e22935f03cc40499eccb7c6a4aac65d : SUCCESS in 44m 13s\n- mistral-devstack-train https://zuul.opendev.org/t/openstack/build/770afb3b9a1e46079b3653658a41e7df : SUCCESS in 40m 57s\n- mistral-devstack-stein https://zuul.opendev.org/t/openstack/build/ee649200c46442618d66847fa82810d8 : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":1},{"id":"aff43e209a575e24ab78874bc34e2b0526937e9c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-11 11:32:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82e782a188debed06a880f04439cff2675731b27","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-12 05:49:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"d0970ac6d173561a6e1f6bff0d41e9dda0b88956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 05:50:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43a3c0bafb7792bc27584a4dc01e17e6c65844c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:09:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"74caf3e7fb954d9dc23d18baec37ee820911ac44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:09:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e598b4813fe44a8ad192c5ddad3c866 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc86804e2c9348eb82c073723ff50f80 : SUCCESS in 3m 47s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9960c53dcd9249b4aef66f3ce2635937 : SUCCESS in 44m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c271f66e72624daba650d9bbdbcb90c9 : SUCCESS in 43m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81942bdfefb84f77915bc5f13044a39d : SUCCESS in 43m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/be9b633ecb8c45fa9c7e8c6c413e4e03 : SUCCESS in 44m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0bf11b4d161c42b9bfdf2eadb2f4c326 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9482b6530d36635667cac83fe566f2b5209abe08","revisions":{"9482b6530d36635667cac83fe566f2b5209abe08":{"kind":"REWORK","_number":1,"created":"2020-05-05 22:16:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/725693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/93/725693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/93/725693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/93/725693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/93/725693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/93/725693/1"}}},"commit":{"parents":[{"commit":"3ff9a7efc4d1127634389eb0e5bdabe3ec63570b","subject":"Setting Python version explicitly to 3 for devstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/3ff9a7efc4d1127634389eb0e5bdabe3ec63570b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:08:09.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:14:42.000000000","tz":-300},"subject":"Add stable branch jobs on the plugins master gate","message":"Add stable branch jobs on the plugins master gate\n\nSupported stable branches in maintenance state\n(stein, train, ussuri) use the Tempest and plugins master\nversion for their testing. To make sure master version (new\ntests or any changes) is compatible with stable branch testing,\nwe need to add the stable branches job on the plugins master gate.\nTempest and many other plugins are already testing the stable\nbranches in same way[1].\n\nUssuri onwards, Tempest or plugins are py3 only, so we are\nrunning only python3 version of stable jobs\n\nTempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\n [1] https://opendev.org/openstack/tempest/src/commit/e8f1876aa6772077f85f380677b30251c2454505/.zuul.yaml#L646\n\nChange-Id: I9498979712375715eab79e77a1c0e89d0efe33b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/9482b6530d36635667cac83fe566f2b5209abe08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/9482b6530d36635667cac83fe566f2b5209abe08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
