)]}'
{"id":"openstack%2Fmistral-tempest-plugin~727517","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I4f7aa67e0663ddea6a5200a55005cf2ce5a0396e","project":"openstack/mistral-tempest-plugin","branch":"master","hashtags":[],"change_id":"I4f7aa67e0663ddea6a5200a55005cf2ce5a0396e","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-13 00:29:39.000000000","updated":"2020-05-13 07:15:51.000000000","submitted":"2020-05-13 07:15:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727517-1589354151458-eb83fbf0","meta_rev_id":"19f2914dfb95e0f4dc5cd564e0193c23409afc53","_number":727517,"virtual_id_number":727517,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 07:15:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-05-13 06:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-13 04:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-05-13 06:08:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 04:52:52.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-05-13 06:08:12.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-05-13 07:15:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44b2f295c6895e20ac3f171874fc532d2298866c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 00:29:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edab8f4fe83eab4646a75de1d28eb6b078228794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 01:56:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77362aa04eae4564b0ca5c28d6338156 : SUCCESS in 3m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f95d8873bc14b77badee9a476bd62b3 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b26637e3af4f638173e704648c7c43 : SUCCESS in 4m 24s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cfa4d8001cb24aa487da81f93e427ae0 : SUCCESS in 43m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57266aba9fdd41d1b892eaba694f7419 : SUCCESS in 42m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82fe9a3556a9490492c5b3ca2e5f2be8 : SUCCESS in 42m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/95261694ad7c458aaaf56f306fefac22 : SUCCESS in 47m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/06d42a358e04458bb20fffb81e08b1f4 : SUCCESS in 45m 15s\n- mistral-devstack-ussuri https://zuul.opendev.org/t/openstack/build/48aa521aa6ac4ba290602621d771840b : SUCCESS in 46m 10s\n- mistral-devstack-train https://zuul.opendev.org/t/openstack/build/4b9a79678b5e478b897516198d1f8d52 : SUCCESS in 44m 37s\n- mistral-devstack-stein https://zuul.opendev.org/t/openstack/build/af722e378da0444c9ec8bbbac30ca201 : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9f823630c16047f54c6de4ebd2e289b79ac26381","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-13 04:52:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f7fc781e8b8807bcc9cd2aa36d81f8947fa5db8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-05-13 06:08:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0cfe81b18286b020521d84ec1861fe21d748e172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 06:08:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3068d2c7809f9673276a427a18807a338819b3c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 07:15:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19f2914dfb95e0f4dc5cd564e0193c23409afc53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 07:15:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3905276d04d44a58bc440470f042cb99 : SUCCESS in 5m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa67a2259a674cd9921f9d01904839eb : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6ac1d08f744a7a99925d1d8c808562 : SUCCESS in 4m 37s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/015a4212c78d4fe78b4f04d4e053c686 : SUCCESS in 57m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/577e70e0c7d24f689d317f97ba97e36f : SUCCESS in 49m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46d6d82581d54765aaf4582e1120c463 : SUCCESS in 46m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9561ac35d49b41b192847b50a0efb691 : SUCCESS in 49m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f4c1a33f684d4789845113a94a2eaabe : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d9545bd41c4547779f4ad40c7ad5ec0cfec4b9c","revisions":{"4d9545bd41c4547779f4ad40c7ad5ec0cfec4b9c":{"kind":"REWORK","_number":1,"created":"2020-05-13 00:29:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/727517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/17/727517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/17/727517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/17/727517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/17/727517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/17/727517/1"}}},"commit":{"parents":[{"commit":"7b4f6b65c94258972fc1ccb3d533985ff0398a56","subject":"Merge \"Add stable branch jobs on the plugins master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/7b4f6b65c94258972fc1ccb3d533985ff0398a56"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-12 23:53:39.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 00:29:38.000000000","tz":-300},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I4f7aa67e0663ddea6a5200a55005cf2ce5a0396e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/4d9545bd41c4547779f4ad40c7ad5ec0cfec4b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/4d9545bd41c4547779f4ad40c7ad5ec0cfec4b9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
