)]}'
{"id":"openstack%2Fmistral-tempest-plugin~791568","triplet_id":"openstack%2Fmistral-tempest-plugin~master~Ifde9c6b18a034e2784459a9dbfb22b080166781e","project":"openstack/mistral-tempest-plugin","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"Ifde9c6b18a034e2784459a9dbfb22b080166781e","subject":"[community goal] Update contributor documentation","status":"MERGED","created":"2021-05-14 22:06:20.000000000","updated":"2021-07-15 04:27:38.000000000","submitted":"2021-07-15 04:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791568-project-ptl-and-contrib-docs","meta_rev_id":"e0a498efdc6fcb9bd55d2d75f6453f917213b899","_number":791568,"virtual_id_number":791568,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 04:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2021-07-06 14:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2021-07-15 04:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-07-07 14:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2021-07-15 04:22:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 22:21:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 07:05:11.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-06-15 22:07:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2021-07-06 07:01:49.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"bffee131415a9340c439fee34f22c73e831cf2fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-14 22:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4570f6b0320349e78491065221f81a8e65064605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 22:21:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a50e0dc87d6401391b73f249d14f878 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad47a4de89a46a697361fa5efc224e5 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5f5e868b3834952dadb11176b02e862ad025f165","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-26 17:52:53.000000000","message":"Patch Set 1: Workflow-1\n\n-W until we migrate to OFTC network and then I will update this contributors guide accordingly. \n\n- http://lists.openstack.org/pipermail/openstack-discuss/2021-May/022718.html","accounts_in_message":[],"_revision_number":1},{"id":"330e946479f76cde6fc0329d903247567e93530a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-31 20:33:39.000000000","message":"Patch Set 1: -Workflow\n\nthis is good to merge","accounts_in_message":[],"_revision_number":1},{"id":"1bf41a9d4008bbaca1e1e545b6fdd06feb853ffe","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-06-01 07:05:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b5908f94705564e0043215328a3a7c676fcf4fe","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-07-06 07:01:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease find review comment.","accounts_in_message":[],"_revision_number":1},{"id":"d94f092798f76c1ec14c493274e3d1aa12b662ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-06 14:03:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"214ea947ca56ce58aa625627e7b465766ae3e532","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-06 14:03:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e5c1c2fcd85e8a4158c339da822e378ce542740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 14:13:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e079040f6a4f06a15d08591f3a26ce : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293b4c01d3804d15bc26a3e7689b0a6d : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"cdff866193d5b3425a9777610e7973bbfa548e40","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-07-06 14:20:56.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for rework. LGTM!!","accounts_in_message":[],"_revision_number":2},{"id":"60cfabfbee82b2dd99537eb9f458fdb84b3056ef","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-07 14:09:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4b16e0c5eff7bacee9da0e9495fd06a7caa95a67","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2021-07-15 04:22:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c9c4bae1926986c3cabf09ba9687a7121838b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 04:22:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0593edee4062bd88366253d73b56b134d24ddbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 04:27:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810c63fabcb147528f9b4aaf0ec71802 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/837a54907c8d4bfe879a33a9f2fda77e : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e0a498efdc6fcb9bd55d2d75f6453f917213b899","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 04:27:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"091926bbb405bf8d0249156f11f090dc455077dc","revisions":{"34f97a455389e4370c2161645e3dea4cbf404e5f":{"kind":"REWORK","_number":1,"created":"2021-05-14 22:06:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/791568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/68/791568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/1"}}},"commit":{"parents":[{"commit":"437e2b680c44a330c1a61106ec117835349d23fd","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/437e2b680c44a330c1a61106ec117835349d23fd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-05-14 22:05:17.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-05-14 22:05:17.000000000","tz":-300},"subject":"[community goal] Update contributor documentation","message":"[community goal] Update contributor documentation\n\nThis patch updates/adds the contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nStory: #2007236\nTask: #38537\nChange-Id: Ifde9c6b18a034e2784459a9dbfb22b080166781e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/34f97a455389e4370c2161645e3dea4cbf404e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/34f97a455389e4370c2161645e3dea4cbf404e5f"}]},"branch":"refs/heads/master"},"091926bbb405bf8d0249156f11f090dc455077dc":{"kind":"REWORK","_number":2,"created":"2021-07-06 14:03:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/791568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/68/791568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/68/791568/2"}}},"commit":{"parents":[{"commit":"437e2b680c44a330c1a61106ec117835349d23fd","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/437e2b680c44a330c1a61106ec117835349d23fd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-05-14 22:05:17.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-07-06 14:03:23.000000000","tz":0},"subject":"[community goal] Update contributor documentation","message":"[community goal] Update contributor documentation\n\nThis patch updates/adds the contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nStory: #2007236\nTask: #38537\nChange-Id: Ifde9c6b18a034e2784459a9dbfb22b080166781e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/091926bbb405bf8d0249156f11f090dc455077dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/091926bbb405bf8d0249156f11f090dc455077dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
