)]}'
{"id":"openstack%2Fmistral-tempest-plugin~922267","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I444bed741852480642e45c2a146898249aacf93e","project":"openstack/mistral-tempest-plugin","branch":"master","topic":"retire-devstack-gate","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2024-10-29 18:06:15.000000000","reason":"Change was abandoned"},"37102":{"account":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"last_update":"2024-10-29 18:06:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I444bed741852480642e45c2a146898249aacf93e","subject":"Drop devstack-gate","status":"ABANDONED","created":"2024-06-19 02:58:17.000000000","updated":"2024-10-29 18:06:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88fa3e7d6635a9371adcac03c33b6ed122df0b99","_number":922267,"virtual_id_number":922267,"owner":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-19 03:50:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-19 03:48:50.000000000","updated_by":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2024-06-19 03:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b65a330eeefcff71d4c87893bdf1442e2df28d1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"date":"2024-06-19 02:58:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d629ba22c77146e621e65b8b54276e8a2bdafe1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 03:50:24.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3efbb7e01bc74eaf83827839f82783f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088318c7272144eab3b3cbf96fa3eedc : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65820ae9993b45eab5355e35959672da : SUCCESS in 2m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b25223ad1bea4e82b7f335257ac281f6 : SUCCESS in 43m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4e1b8db9f4f4cf8858a804f65266484 : SUCCESS in 44m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a08dc56fcbb742d3a15e9bb9e43b66a4 : SUCCESS in 42m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/232ee2e35a394102bfb87eba49f1965b : SUCCESS in 47m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f92b9b61080641faa5ec53a37906c421 : SUCCESS in 51m 17s\n- mistral-devstack-ussuri https://zuul.opendev.org/t/openstack/build/9d7008bda8bd485595bb5ca4d0b6fe87 : FAILURE in 15m 07s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/mistral-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"branches\" attribute\nof the following project stanza:\n\n  ...\n          - mistral-devstack\n          - mistral-devstack-tempest-ipv6-only\n          - mistral-devstack-non-apache-tempest-ipv6-only\n          - mistral-devstack-non-apache:\n              branches: ^(?!stable/(newton|ocata)).*$\n          - mistral-devstack-kombu:\n              branches: ^(?!stable/newton).*$\n          - mistral-devstack-ussuri\n      queue: mistral\n  ...\n\n  in \"openstack/mistral-tempest-plugin/.zuul.yaml@master\", line 88","accounts_in_message":[],"_revision_number":1},{"id":"88fa3e7d6635a9371adcac03c33b6ed122df0b99","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-29 18:06:15.000000000","message":"Abandoned\n\nalready done in 921822","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc331062067e412e5e455a6bf94d0d36fa2dc15b","revisions":{"cc331062067e412e5e455a6bf94d0d36fa2dc15b":{"kind":"REWORK","_number":1,"created":"2024-06-19 02:58:17.000000000","uploader":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"ref":"refs/changes/67/922267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/67/922267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/67/922267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/67/922267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/67/922267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/67/922267/1"}}},"commit":{"parents":[{"commit":"2f92367896338ec38205f2486c6339a96c21604c","subject":"Merge \"Wait till execution is success for test_task_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/2f92367896338ec38205f2486c6339a96c21604c"}]}],"author":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-06-19 02:56:28.000000000","tz":-240},"committer":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-06-19 02:56:28.000000000","tz":-240},"subject":"Drop devstack-gate","message":"Drop devstack-gate\n\nThe devstack-gate repo has been retired, let\u0027s drop all references\nto it to avoid errors like this in Zuul:\n\nThe project \"openstack/devstack-gate\" was not found.\n\nChange-Id: I444bed741852480642e45c2a146898249aacf93e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/cc331062067e412e5e455a6bf94d0d36fa2dc15b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/cc331062067e412e5e455a6bf94d0d36fa2dc15b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
