)]}'
{"id":"openstack%2Fmistral-tempest-plugin~933357","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I3282762608b97c99c8600e6a55e09898103b49d0","project":"openstack/mistral-tempest-plugin","branch":"master","topic":"support-py312","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-10-31 12:37:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3282762608b97c99c8600e6a55e09898103b49d0","subject":"Support py3.12 and drop py3.8","status":"MERGED","created":"2024-10-24 19:20:19.000000000","updated":"2024-10-31 12:37:15.000000000","submitted":"2024-10-31 12:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933357-support-py312","meta_rev_id":"9eb74c5527e3c1eb57393713c03351688623bd05","_number":933357,"virtual_id_number":933357,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 12:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-31 11:01:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-31 11:01:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 20:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 11:00:41.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2024-10-31 11:01:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"ee4ce5761941f37b42875b0384627a9d655aebec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-24 19:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa406bf795ecd6603dba409b7eb98518ecd64409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 20:13:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69f9d1bf6d2a4aea8a48dcf46fc903f6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b65ca467e2e4c7887947fb00abe8c0a : SUCCESS in 1m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e3a073a95b841fdb8d5b86e6068a30d : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b9273f1c8d4c8394fb3c1ced093b1a : SUCCESS in 4m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5daccca3315a44ac820bbcd541e8fc0b : FAILURE in 25m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28f78f1b9a4c4ff2b26919c92e7bfa19 : FAILURE in 40m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54a47e089f3c4a6f946f1dadd27f819d : FAILURE in 42m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cb496083710c424ea420a20755c5fc4b : FAILURE in 23m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/badaa339a43b4fd693ef0dcca8a4de72 : FAILURE in 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ea44480d1f8c9c1241d6c8125f370754a1255e5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-25 22:25:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3a15b2aa91f0ad35c6479feb04dbc37f971f7a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 23:09:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d2065e614ed47e58a950d22de7006b2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/828a3159ffb041cdb154f7e12eea13d0 : SUCCESS in 1m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b67f4d0e2b7498496876bd96fbb9c14 : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11eb0683b4294b5ab86b08a091153b94 : SUCCESS in 1m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7fe7bc9aeb9b4e1d9453a3784319c5d1 : FAILURE in 28m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a718899e79544959a4ef0199ce56356c : FAILURE in 42m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd612fd42d5644fb91f13e87a32ae92f : FAILURE in 38m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0c8e532f284a4dbfbd1fb29acf181b45 : FAILURE in 20m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6d2835db9fe34129b2b78dd9dde27cd3 : FAILURE in 26m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a407dbd01314fcf5b2b6b0113b7f5c8a5f27b9cc","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-31 11:00:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"faa257e78fd0c5e8b934330aa4e79413a5db4572","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-10-31 11:01:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8a5f606035b891611dab185e2d8b104733b7d8b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 11:59:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bda755295904fc998ec50dfed3c83b5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ceb4954436c44209ede2a1575cc45d1 : SUCCESS in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392bac9b6ac3416ea08e10f4d4ec9535 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b7e42ff31146508f73317337003ac4 : SUCCESS in 3m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2e9a93ba93004d0984dd5b9b729cedab : SUCCESS in 27m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0cf5ad7cb104035aa2d31e470d5e5b6 : SUCCESS in 43m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7a5f30923a24eeab8079ea52a37c259 : SUCCESS in 34m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/eb03ba91c5584f4681c5f7cb2eebfe36 : SUCCESS in 43m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0b10b21229bd4da2be2aaf9b56690d9b : SUCCESS in 28m 04s","accounts_in_message":[],"_revision_number":2},{"id":"047c27594f0cc04067b2299f66f0a934b3262be4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 11:59:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"969d6bde6127eb87c3ab3bd87ac9d9fcc845a3da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:37:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7689f7cb5df44388b6b7f8e6b7c78ba\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2611d217e2e940ef93f06e033f7d570f : SUCCESS in 1m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd555938c7449c1b8da4ddbcfa71ab4 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0be22cc11c24ecda6c3e270777e4734 : SUCCESS in 3m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0265d5d7a8634f308d1d47fa37481f44 : SUCCESS in 33m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c690e8126f7743cfb708b583e8c675ca : SUCCESS in 35m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5641b39f6f7b44e88380c7e7764bc9c1 : SUCCESS in 35m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2c1d17dedb8847f0925db89a7688ff7f : SUCCESS in 29m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/342c4ca59e264b958e585bea9bffdb81 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9eb74c5527e3c1eb57393713c03351688623bd05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:37:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfb90c512782a50b8536c0352b35365c00b8808f","revisions":{"ef798b2d35ded46986a9019a855e97d84e0ab2bc":{"kind":"REWORK","_number":1,"created":"2024-10-24 19:20:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/57/933357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/57/933357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/1"}}},"commit":{"parents":[{"commit":"fab1c9d10a10164956ec3ec54d4d001571ff2744","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/fab1c9d10a10164956ec3ec54d4d001571ff2744"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:01:47.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:01:47.000000000","tz":-420},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nIn tempest plugin, we test from python 3.8 till\npython 3.12 so updating the same in python\nclassifier in setup.cfg\n\nChange-Id: I3282762608b97c99c8600e6a55e09898103b49d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/ef798b2d35ded46986a9019a855e97d84e0ab2bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/ef798b2d35ded46986a9019a855e97d84e0ab2bc"}]},"branch":"refs/heads/master"},"dfb90c512782a50b8536c0352b35365c00b8808f":{"kind":"REWORK","_number":2,"created":"2024-10-25 22:25:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/57/933357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/57/933357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/57/933357/2"}}},"commit":{"parents":[{"commit":"fab1c9d10a10164956ec3ec54d4d001571ff2744","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/fab1c9d10a10164956ec3ec54d4d001571ff2744"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-24 19:01:47.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-25 22:15:43.000000000","tz":-420},"subject":"Support py3.12 and drop py3.8","message":"Support py3.12 and drop py3.8\n\nAs per the current release tested runtime, we test\nfrom python 3.9 till python 3.12 so updating the same\nin python classifier in setup.cfg\n\nAlso, dropping the python3.8 support\n- https://lists.openstack.org/archives/list/openstack-discuss@li\u003e\n\nChange-Id: I3282762608b97c99c8600e6a55e09898103b49d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/dfb90c512782a50b8536c0352b35365c00b8808f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/dfb90c512782a50b8536c0352b35365c00b8808f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
