)]}'
{"id":"openstack%2Fmistral-tempest-plugin~947309","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I05265d80fcf96628ef313c08b255e9659b088ea6","project":"openstack/mistral-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-16 07:08:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05265d80fcf96628ef313c08b255e9659b088ea6","subject":"Remove unnecessary +x mode","status":"MERGED","created":"2025-04-16 03:11:35.000000000","updated":"2025-04-16 07:08:42.000000000","submitted":"2025-04-16 07:08:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947309","meta_rev_id":"6fc2bf2324172827b657e7650bdeb06cea5c7fc1","_number":947309,"virtual_id_number":947309,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-16 07:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-16 06:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-16 06:30:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 03:57:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 06:30:12.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"477cf5f411522692fa6ba59b060a2fd196486574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-16 03:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"413e052ea85f8a27ed67fac6bbe24a0242daf4e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 03:57:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0afd2b2e39cd41009b1dd30bfe79b57a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da2840bf30f416b9c08db313231c27e : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be24386ea0f4418d9c28fd3227957a95 : SUCCESS in 1m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b8d7b3052b0b4d79bb729a964960f571 : SUCCESS in 38m 29s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/6ce3fdcb725344e793d3c3efce334213 : SUCCESS in 37m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f66c0600c8c14825b14202644917e4e4 : SUCCESS in 20m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10941f1942384a77a6fd8ddb664db5ae : SUCCESS in 26m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/56f73580d393421e80713db54cdb77cc : SUCCESS in 31m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2ca82eafc10f4a66b293596198bc58ec : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":1},{"id":"178e4a84d01a541d119c3363779d7d02ac9f6d26","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-16 06:30:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da563d202db46e0d183c30a6d7830cbbca041755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 06:30:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a1eaca504d8ae583bdc738eebc627987f4a1bc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 07:08:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ea121932d94e1998ce128231b2f3d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c92b7647a2e41a7bc3a3ceec18e0bc0 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba97a33558247b8b6b6366d0c133c9b : SUCCESS in 1m 45s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4fc9db26319344d0b7941b710ccb6201 : SUCCESS in 19m 48s\n- mistral-devstack-jammy https://zuul.opendev.org/t/openstack/build/f3aab3cb09b34106a8bdac83ad678f88 : SUCCESS in 14m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8d4776a912c40b6b84687d27c0fcfca : SUCCESS in 27m 40s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff61e2e1a6bc4abe8355e60ce12e6d11 : SUCCESS in 28m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/aa59b43f2cb841109ec36d779a7f21cc : SUCCESS in 27m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3faef4a0b2144dc3965aee62d470c1a8 : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6fc2bf2324172827b657e7650bdeb06cea5c7fc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 07:08:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bf69040b4578fc929ea9e1448a7a786298789e0","revisions":{"2bf69040b4578fc929ea9e1448a7a786298789e0":{"kind":"REWORK","_number":1,"created":"2025-04-16 03:11:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/947309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/09/947309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/09/947309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/09/947309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/09/947309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/09/947309/1"}}},"commit":{"parents":[{"commit":"a707405e33d81ae797103dccf2a9de9df13276cc","subject":"Skip testing v2/services endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/a707405e33d81ae797103dccf2a9de9df13276cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 03:11:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 03:11:20.000000000","tz":540},"subject":"Remove unnecessary +x mode","message":"Remove unnecessary +x mode\n\nThese files are not actually executable.\n\nChange-Id: I05265d80fcf96628ef313c08b255e9659b088ea6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/2bf69040b4578fc929ea9e1448a7a786298789e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/2bf69040b4578fc929ea9e1448a7a786298789e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
