)]}'
{"id":"openstack%2Fmistral-tempest-plugin~955335","triplet_id":"openstack%2Fmistral-tempest-plugin~master~Ie90a4f23ba598707f315bc858f61e50cb7dcb4d3","project":"openstack/mistral-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-20 12:00:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie90a4f23ba598707f315bc858f61e50cb7dcb4d3","subject":"Drop removed devstack parameters","status":"MERGED","created":"2025-07-18 02:21:14.000000000","updated":"2025-07-20 12:00:10.000000000","submitted":"2025-07-20 12:00:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955335","meta_rev_id":"ce608a0a9e90a2fb914a135d3f4bf3b013a57378","_number":955335,"virtual_id_number":955335,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-20 12:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-18 13:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-18 13:59:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-18 03:11:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-18 13:59:03.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"db79f829d186bf872375c984a87954ccc272fb46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-18 02:21:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55cbbf998e29c871ddec2c0f188c7d6bb6fd46e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-18 02:25:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f494a9770d7acc6d612ddeafa2894ec8c8a44aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 03:11:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30252c060fcc41db9e40a6909ec554bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df12eb9b9cf244488d0197bb895d7b4b : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4243599452834876a39a9a47087d095c : SUCCESS in 3m 22s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d176fb8d68ec4bb6a6d01ce221bc15a1 : SUCCESS in 38m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1580b72d6c934652affda5ff45e6fb18 : SUCCESS in 40m 44s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8306208f1933477aa942f988a3a10fcd : SUCCESS in 39m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1edcea2ac8c04ae5ba8c0a8110b99b00 : SUCCESS in 36m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/632c5f3047bc4fae8e45608c1729ac97 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6c9bd84c3a6544553fbdf99561915207ce7b4a2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-18 05:30:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e2890608077d634147d35ea63db3b036185cf3fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 05:53:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6112bf1827ad481f83b40ec2fbb8acbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87875fdb75c4376928ca18606526ed4 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31dcc7ad5a31485188acb498c16b48db : SUCCESS in 2m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/51b21fe759c14124a23e2beae4606c07 : SUCCESS in 20m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/626c08d14e3742609bb30dfd8d530058 : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":3},{"id":"136eb992d04affd86260c6ca8c480110a12a202b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-07-18 13:59:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"147f574b7f81353b816540f9687c87a6dceb36f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 11:12:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb549f3a34dc4918409c3532279d93eaa2ab6d0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 12:00:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165f5c0ccd1b497ca6323c71fe1a2070\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e496917e321d43178438e7e4c05bdbdf : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4010c0d2f024525a5ec52420910693f : SUCCESS in 1m 55s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/805652dc5deb45c5a4d0a7541acd211d : SUCCESS in 36m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19786d472bb4421381b670b4eadceb31 : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":3},{"id":"ce608a0a9e90a2fb914a135d3f4bf3b013a57378","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 12:00:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"05d4ca892ba9724ec4dbee335765d7ba7f69ba71","revisions":{"9d8f41cae57ba7982c93baf42fe0ec7277380408":{"kind":"REWORK","_number":1,"created":"2025-07-18 02:21:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/955335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/35/955335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/1"}}},"commit":{"parents":[{"commit":"23c64bee7f79679a745bad670fd4011a168aae12","subject":"Zuul: do not use USE_PYTHON3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/23c64bee7f79679a745bad670fd4011a168aae12"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-18 02:20:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-18 02:20:49.000000000","tz":540},"subject":"Drop MISTRAL_USE_MOD_WSGI","message":"Drop MISTRAL_USE_MOD_WSGI\n\n... because the parameter was removed.\n\nNote that this parameter defaults to True in stable branches so\nremoving the override does not affect stable branches.\n\nDepends-on: https://review.opendev.org/954952\nChange-Id: Ie90a4f23ba598707f315bc858f61e50cb7dcb4d3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/9d8f41cae57ba7982c93baf42fe0ec7277380408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/9d8f41cae57ba7982c93baf42fe0ec7277380408"}]},"branch":"refs/heads/master"},"1b22a3d4d4da7a94545ca4f0417c448682614a66":{"kind":"REWORK","_number":2,"created":"2025-07-18 02:25:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/955335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/35/955335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/2"}}},"commit":{"parents":[{"commit":"23c64bee7f79679a745bad670fd4011a168aae12","subject":"Zuul: do not use USE_PYTHON3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/23c64bee7f79679a745bad670fd4011a168aae12"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-18 02:20:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-18 02:25:42.000000000","tz":540},"subject":"Remove removed devstack parameters","message":"Remove removed devstack parameters\n\nMISTRAL_USE_MOD_WSGI and MISTRAL_RPC_IMPLEMENTATION were removed from\ndevstack plugin.\n\nNote\n* MISTRAL_USE_MOD_WSGI defaults to True in stable branches so removing\n  the override does not affect stable branches.\n* MISTRAL_RPC_IMPLEMENTATION defaults to oslo in all open stable\n  branches so removing the override does not affect any branches,\n  including master.\n\nDepends-on: https://review.opendev.org/954952\nChange-Id: Ie90a4f23ba598707f315bc858f61e50cb7dcb4d3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/1b22a3d4d4da7a94545ca4f0417c448682614a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/1b22a3d4d4da7a94545ca4f0417c448682614a66"}]},"branch":"refs/heads/master"},"05d4ca892ba9724ec4dbee335765d7ba7f69ba71":{"kind":"REWORK","_number":3,"created":"2025-07-18 05:30:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/955335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/35/955335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/35/955335/3"}}},"commit":{"parents":[{"commit":"23c64bee7f79679a745bad670fd4011a168aae12","subject":"Zuul: do not use USE_PYTHON3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/23c64bee7f79679a745bad670fd4011a168aae12"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-18 02:20:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-18 05:30:41.000000000","tz":540},"subject":"Drop removed devstack parameters","message":"Drop removed devstack parameters\n\nMISTRAL_USE_MOD_WSGI and MISTRAL_RPC_IMPLEMENTATION were removed from\ndevstack plugin.\n\nNote\n* MISTRAL_USE_MOD_WSGI defaults to True in stable branches so removing\n  the override does not affect stable branches.\n* MISTRAL_RPC_IMPLEMENTATION defaults to oslo in all open stable\n  branches so removing the override does not affect any branches,\n  including master.\n\nDepends-on: https://review.opendev.org/954952\nChange-Id: Ie90a4f23ba598707f315bc858f61e50cb7dcb4d3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/05d4ca892ba9724ec4dbee335765d7ba7f69ba71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/05d4ca892ba9724ec4dbee335765d7ba7f69ba71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
