)]}'
{"id":"openstack%2Fmistral-tempest-plugin~959561","triplet_id":"openstack%2Fmistral-tempest-plugin~master~I80acf0c82e0ba46059d8f47f4a762fd828dc5f6d","project":"openstack/mistral-tempest-plugin","branch":"master","hashtags":[],"change_id":"I80acf0c82e0ba46059d8f47f4a762fd828dc5f6d","subject":"Switch from httpbin to httpbingo","status":"MERGED","created":"2025-09-04 16:44:52.000000000","updated":"2025-09-04 18:13:35.000000000","submitted":"2025-09-04 18:13:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959561","meta_rev_id":"727625a42aafbf7eb727b8feff34457b9b36b8a5","_number":959561,"virtual_id_number":959561,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 18:13:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 17:35:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 17:35:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 17:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2ee50bf37862f778d75b778240206bc902d0403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 16:44:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d05bfd89604dbce511e854768e9e36ba55fbfeb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 17:26:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ef9864aeaa24a209b056b2a5f2742b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df71ccd469e4dbdb42e382a1396a2aa : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e50ab71aa4497facc46b1ac8b5ae0b : SUCCESS in 1m 53s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/cd634129b9b748e88de726c5e5e4c8a3 : SUCCESS in 35m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b32d8c1109de416088be12dd2c456fb4 : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":1},{"id":"428650e681f05b6e7f3f464f9306c1b470ad0a80","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 17:35:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0547b0b312e80d4cd659b7e9ff2a1486737340dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 17:36:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac25119a9f0201a2493435c486735f8538571908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:13:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e1973fe176645cc840f366fc2a768d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1fd40d4b044448a84eab36c587b4ea3 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01521de500e45b487bbd9cfb23e8ec2 : SUCCESS in 2m 26s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/05ebe52f01ff4d7383fb432f3b42e8d3 : SUCCESS in 20m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb2af14257a140d4b3c8cfc9199d5964 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"727625a42aafbf7eb727b8feff34457b9b36b8a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:13:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e948a8433143d06b2821ecb26a1a46ed916baa2d","revisions":{"e948a8433143d06b2821ecb26a1a46ed916baa2d":{"kind":"REWORK","_number":1,"created":"2025-09-04 16:44:52.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/61/959561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/61/959561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/61/959561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/61/959561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/61/959561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/61/959561/1"}}},"commit":{"parents":[{"commit":"cc5ac976b35376788c5787ab054b3b81a964821a","subject":"Skip test that rely on httpbin.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/cc5ac976b35376788c5787ab054b3b81a964821a"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-04 16:43:04.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-04 16:43:04.000000000","tz":120},"subject":"Switch from httpbin to httpbingo","message":"Switch from httpbin to httpbingo\n\nIt seems a nice remplacement\n\nChange-Id: I80acf0c82e0ba46059d8f47f4a762fd828dc5f6d\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/e948a8433143d06b2821ecb26a1a46ed916baa2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/e948a8433143d06b2821ecb26a1a46ed916baa2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
