)]}'
{"id":"openstack%2Fmistral-tempest-plugin~987998","triplet_id":"openstack%2Fmistral-tempest-plugin~master~Ic209d9eaaad4c86b3851b887907b2fd25b9f3e77","project":"openstack/mistral-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-11 23:02:42.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-05-11 22:09:30.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ic209d9eaaad4c86b3851b887907b2fd25b9f3e77","subject":"Drop remaining reference to Heat CloudWatch API","status":"MERGED","created":"2026-05-10 14:40:36.000000000","updated":"2026-05-11 23:02:42.000000000","submitted":"2026-05-11 23:02:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987998","meta_rev_id":"8c2fe21ed9f5e35d10fbc1e09190fa2eaf8b6f45","_number":987998,"virtual_id_number":987998,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 23:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-11 22:07:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-05-11 22:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-05-11 22:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-10 15:22:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 22:07:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-11 22:07:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"c350fd27b2d449c145041af7a52d9012b5e8145f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 14:40:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"602a25f8b067a7d0770e57ae91a1ab0789c4c5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 15:22:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64ceff9425f94028997671a18a0271a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b25e3ab55c41f8a9ee4c1f1564ba37 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7086b4dacb5547f9a91555ba293c4fb0 : SUCCESS in 2m 18s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c20d8c1d4d4c44379edebef567361779 : FAILURE in 32m 31s\n- mistral-devstack-2026-1 https://zuul.opendev.org/t/openstack/build/219169d92de448a9acf0419c5934fe8d : FAILURE in 34m 32s\n- mistral-devstack-2025-2 https://zuul.opendev.org/t/openstack/build/14657df7a2034013867258b3c165dc0c : FAILURE in 31m 24s\n- mistral-devstack-2025-1 https://zuul.opendev.org/t/openstack/build/cd18a5fdecac4375aa9a8e3ea38af50d : FAILURE in 16m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9859059be4bd46bd9e18a14edf735b2d : FAILURE in 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"53d0337be0eadf611020a3d55b7b8bf3dfc93f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 15:23:20.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1a9201b97cc941810ec3d7b51aa31e8574ceaa52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 16:03:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a8db3dbb7c14b8b9178e0ad723dd144\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa302387fa524b60b6b6e8035fee6d1e : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc29ee701ae48c9bdca3305846bd2e9 : SUCCESS in 2m 06s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/48dc3b640fd94b3fa49e89325c08003e : FAILURE in 32m 25s\n- mistral-devstack-2026-1 https://zuul.opendev.org/t/openstack/build/d44e06ea21a54c3e9a0ec795c7d51558 : FAILURE in 33m 27s\n- mistral-devstack-2025-2 https://zuul.opendev.org/t/openstack/build/ca4a5edc8d3e4a229e20e256cdaf29cc : FAILURE in 31m 40s\n- mistral-devstack-2025-1 https://zuul.opendev.org/t/openstack/build/6f98ecbfb032478985bb88953e217ca9 : FAILURE in 32m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34060bad051e4cc0a5004fea671b7a7c : FAILURE in 32m 02s","accounts_in_message":[],"_revision_number":2},{"id":"2338b0d6e1357b054e9986f5e51fd8562e9b9189","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-11 16:12:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a0face4c08be3e18bc0bbc9f553f290714901b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 16:44:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f1ebba3d8b04565bedf22f5d0af00d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d1ab47fa1d45be8efa3cb58a089765 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1975869e9c8e431c962beb05a4fde02d : SUCCESS in 2m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2c68531a12f84d63822a8c498c5eff1c : SUCCESS in 29m 33s\n- mistral-devstack-2026-1 https://zuul.opendev.org/t/openstack/build/f01285c66b7948928160aff9e21b3f0c : SUCCESS in 24m 39s\n- mistral-devstack-2025-2 https://zuul.opendev.org/t/openstack/build/da73a04bcb1f4360b79809176fb10440 : SUCCESS in 28m 28s\n- mistral-devstack-2025-1 https://zuul.opendev.org/t/openstack/build/5e88661f770a42e5a1a35fad34cd98e4 : FAILURE in 12m 00s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36a66180fbc646d49cf9d39676562f1d : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":2},{"id":"fee659be751874e4d9983b607b5eb880ddd15f3f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-11 18:07:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5699e2247e522ed8798340ea1d83b4f7c76bdb8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 19:03:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aa5c2724dc94cb3b4a98f4646d207cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a26a6d4658493b837a237166239683 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fe58db6815498ba5120b3bf382265b : SUCCESS in 2m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/62fd330abe24440a921a416ccadcaa35 : SUCCESS in 30m 49s\n- mistral-devstack-2026-1 https://zuul.opendev.org/t/openstack/build/53ff114d7bf244f0b757a218c2377150 : SUCCESS in 41m 45s\n- mistral-devstack-2025-2 https://zuul.opendev.org/t/openstack/build/671d371b1b7c45158b3ac50f523a9e1c : SUCCESS in 31m 40s\n- mistral-devstack-2025-1 https://zuul.opendev.org/t/openstack/build/051ac939b49b4b3e84ba47693502348c : SUCCESS in 50m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/174de6446c1f46e6b766b947de7393a5 : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":2},{"id":"df637f01a86175e65bb9752053b19b50f849c374","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-11 22:07:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1c436d11dbb2e0e56f992853b7f2a48cc60a1a90","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-11 22:09:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"858802ba7d58721e3755052cb6daf388f53d6dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 22:10:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63dc6b9fabc166a86a475ab6203072e355e31df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:02:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7173ba754c3344a8b7855fa8140f38e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e2bc34ed53462cb77d232a0f448fe1 : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e097f3fd0b34ffba29568ddf7081adb : SUCCESS in 2m 17s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/017de99cc9864055806ef1d8cfd62a0f : SUCCESS in 16m 43s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd5bdf5bf8634032aedb83191c590347 : SUCCESS in 24m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8c2fe21ed9f5e35d10fbc1e09190fa2eaf8b6f45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:02:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ecbb3beb9ebfc9d7e9fad0df42c55bd7307e5cb","revisions":{"64d25841e8b33c191ee81d7d003177e50da6adc3":{"kind":"REWORK","_number":1,"created":"2026-05-10 14:40:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/987998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/98/987998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/1"}}},"commit":{"parents":[{"commit":"0a9c9c16d622ca828c6eee1bc2c5181ab04ffaa1","subject":"Merge \"zuul: Use ansible variable to configure tempest plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/0a9c9c16d622ca828c6eee1bc2c5181ab04ffaa1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 14:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 14:37:54.000000000","tz":540},"subject":"Drop remaining reference to Heat CloudWatch API","message":"Drop remaining reference to Heat CloudWatch API\n\nThe service was removed multiple years ago.\n\nAlso remove the \u0027heat\u0027 service which has no effect.\n\nChange-Id: Ic209d9eaaad4c86b3851b887907b2fd25b9f3e77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/64d25841e8b33c191ee81d7d003177e50da6adc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/64d25841e8b33c191ee81d7d003177e50da6adc3"}]},"branch":"refs/heads/master"},"0ecbb3beb9ebfc9d7e9fad0df42c55bd7307e5cb":{"kind":"REWORK","_number":2,"created":"2026-05-10 15:23:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/987998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/98/987998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/98/987998/2"}}},"commit":{"parents":[{"commit":"0a9c9c16d622ca828c6eee1bc2c5181ab04ffaa1","subject":"Merge \"zuul: Use ansible variable to configure tempest plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/0a9c9c16d622ca828c6eee1bc2c5181ab04ffaa1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 14:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 15:23:20.000000000","tz":0},"subject":"Drop remaining reference to Heat CloudWatch API","message":"Drop remaining reference to Heat CloudWatch API\n\nThe service was removed multiple years ago.\n\nAlso remove the \u0027heat\u0027 service which has no effect.\n\nChange-Id: Ic209d9eaaad4c86b3851b887907b2fd25b9f3e77\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/0ecbb3beb9ebfc9d7e9fad0df42c55bd7307e5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/0ecbb3beb9ebfc9d7e9fad0df42c55bd7307e5cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
