)]}'
{"id":"openstack%2Fmistral-tempest-plugin~987999","triplet_id":"openstack%2Fmistral-tempest-plugin~master~Ib4bbdb59b0565ce1af0f95ebac4af0ebda25fb4d","project":"openstack/mistral-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-11 23:02:43.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-05-11 22:09:09.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ib4bbdb59b0565ce1af0f95ebac4af0ebda25fb4d","subject":"Drop override of services enabled in base job","status":"MERGED","created":"2026-05-10 14:40:36.000000000","updated":"2026-05-11 23:02:43.000000000","submitted":"2026-05-11 23:02:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987999","meta_rev_id":"363b63db8e33be2ac370c1a949498a55f405cb38","_number":987999,"virtual_id_number":987999,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 23:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-11 22:06:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-05-11 22:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-05-11 22:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-10 16:04:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 22:06:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-11 22:06:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"314e425521459a8f4ba09ed35e87d3cbfa2b4fd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 14:40:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"836eaa6fb70347f49b11116b006a2901d942a534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 15:23:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0451c347b5b9331a7158c4c60bc16535136f53da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 16:04:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13549404bcf54896bc66b43f5894410a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96c3bbcb99a4c519f50a3ae4dbf0fbd : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a3f367f1874c43bc8aad3a6cbb241d : SUCCESS in 2m 02s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ec13e2d24e774c159455914fe9abbad2 : FAILURE in 31m 33s\n- mistral-devstack-2026-1 https://zuul.opendev.org/t/openstack/build/d31539acd9164cecae45ca0204c9ceff : FAILURE in 32m 49s\n- mistral-devstack-2025-2 https://zuul.opendev.org/t/openstack/build/b84a065499c64c6d9a32f5c66ee5af24 : FAILURE in 32m 05s\n- mistral-devstack-2025-1 https://zuul.opendev.org/t/openstack/build/b23a82b568584f56a2805f38ba1eda45 : FAILURE in 33m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26a0c53ad2a94a4a88ce6cba7187c5f4 : FAILURE in 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f76b9ba54bc0dc1adae77d35ca20d95f44a5e478","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-11 16:12:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e07f2fab61a53f6949cf0719bfec50219ab287fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 17:00:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2da7c7ffe6c464a9c03e52d00700585\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ac4cc035af444da25dd954c54bfdf9 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bfa2a04a09d4012a3489f19c2438d7c : SUCCESS in 2m 27s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f508208101bc40639ed9bfdf28020dab : SUCCESS in 28m 55s\n- mistral-devstack-2026-1 https://zuul.opendev.org/t/openstack/build/dbd0a2d3e1bc4e29b95b1559d925a3a5 : SUCCESS in 22m 37s\n- mistral-devstack-2025-2 https://zuul.opendev.org/t/openstack/build/0b017725dc05454e976d1d7e7ac33e41 : SUCCESS in 40m 32s\n- mistral-devstack-2025-1 https://zuul.opendev.org/t/openstack/build/6f224f68dc4a4230ad71aacb1c4e50f6 : SUCCESS in 39m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4430531473684c3faa5bd78f854bb6dc : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":2},{"id":"e7501f14b93e3031b7d600fb52fac91a70a2ef9a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-11 22:06:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c53c72025139bb815623429baa03b2fead001607","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-11 22:09:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"172f731fed12f781895fbc96543488e5bfbc89a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 22:10:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb4790c5227972bcde9737cae430bafe3fd3826c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:02:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7da136d7bbc437fa28f595401475444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d58f523250fb4eda99e8e9d5c7e4460f : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd0105e5d904da5b831448475acf388 : SUCCESS in 2m 20s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7303de01f587402c8ce6fcd66b5f0502 : SUCCESS in 39m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7f95f8b5a0441609cc0eac7682755c4 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"363b63db8e33be2ac370c1a949498a55f405cb38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:02:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05581dda964ad83f1e72aa4cce16721398a4418b","revisions":{"b2bd3be088dd68a911802f991c97508ad921f694":{"kind":"REWORK","_number":1,"created":"2026-05-10 14:40:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/987999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/99/987999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/1"}}},"commit":{"parents":[{"commit":"64d25841e8b33c191ee81d7d003177e50da6adc3","subject":"Drop remaining reference to Heat CloudWatch API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/64d25841e8b33c191ee81d7d003177e50da6adc3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 14:38:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 14:40:10.000000000","tz":540},"subject":"Drop override of services enabled in base job","message":"Drop override of services enabled in base job\n\nSwift and tempest are both enabled in the devstack-tempest job.\n\nAlso remove a few redundant overrides.\n\nChange-Id: Ib4bbdb59b0565ce1af0f95ebac4af0ebda25fb4d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/b2bd3be088dd68a911802f991c97508ad921f694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/b2bd3be088dd68a911802f991c97508ad921f694"}]},"branch":"refs/heads/master"},"05581dda964ad83f1e72aa4cce16721398a4418b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-10 15:23:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/987999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral-tempest-plugin","ref":"refs/changes/99/987999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral-tempest-plugin refs/changes/99/987999/2"}}},"commit":{"parents":[{"commit":"0ecbb3beb9ebfc9d7e9fad0df42c55bd7307e5cb","subject":"Drop remaining reference to Heat CloudWatch API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/0ecbb3beb9ebfc9d7e9fad0df42c55bd7307e5cb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 14:38:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 15:23:39.000000000","tz":0},"subject":"Drop override of services enabled in base job","message":"Drop override of services enabled in base job\n\nSwift and tempest are both enabled in the devstack-tempest job.\n\nAlso remove a few redundant overrides.\n\nChange-Id: Ib4bbdb59b0565ce1af0f95ebac4af0ebda25fb4d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/05581dda964ad83f1e72aa4cce16721398a4418b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral-tempest-plugin/commit/05581dda964ad83f1e72aa4cce16721398a4418b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"b2bd3be088dd68a911802f991c97508ad921f694","theirs":"0ecbb3beb9ebfc9d7e9fad0df42c55bd7307e5cb","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
