)]}'
{"id":"openstack%2Fmistral~234587","triplet_id":"openstack%2Fmistral~master~I226e8f8c6afbbc94c6348a17099f02b9cf7907a8","project":"openstack/mistral","branch":"master","topic":"bp/mistral-py3","hashtags":[],"change_id":"I226e8f8c6afbbc94c6348a17099f02b9cf7907a8","subject":"Add decode() function for string comparison.","status":"MERGED","created":"2015-10-14 06:23:07.000000000","updated":"2015-10-22 09:05:53.000000000","submitted":"2015-10-22 09:05:51.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3c45c81975938eaa545f4907b99b36f5ff3da212","_number":234587,"virtual_id_number":234587,"owner":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2015-10-22 09:05:51.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":8358,"name":"zhangguoqing","email":"zhang.guoqing@99cloud.net","username":"zhangguoqing"},{"value":0,"_account_id":5558,"name":"Winson Chan","email":"wcchan@stackstorm.com","username":"winson-c-chan"},{"value":0,"_account_id":9432,"name":"Dmitri Zimine","email":"dz@stackstorm.com","username":"dzimine"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":18460,"name":"Jaiveek Shah","email":"jaiveek.shah@gmail.com","username":"JaiveekShah"},{"value":0,"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-10-22 08:55:36.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":8358,"name":"zhangguoqing","email":"zhang.guoqing@99cloud.net","username":"zhangguoqing"},{"value":0,"_account_id":5558,"name":"Winson Chan","email":"wcchan@stackstorm.com","username":"winson-c-chan"},{"value":0,"_account_id":9432,"name":"Dmitri Zimine","email":"dz@stackstorm.com","username":"dzimine"},{"value":2,"date":"2015-10-22 05:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":1,"date":"2015-10-22 05:43:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18460,"name":"Jaiveek Shah","email":"jaiveek.shah@gmail.com","username":"JaiveekShah"},{"value":0,"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-10-22 08:55:36.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":8358,"name":"zhangguoqing","email":"zhang.guoqing@99cloud.net","username":"zhangguoqing"},{"value":0,"_account_id":5558,"name":"Winson Chan","email":"wcchan@stackstorm.com","username":"winson-c-chan"},{"value":0,"_account_id":9432,"name":"Dmitri Zimine","email":"dz@stackstorm.com","username":"dzimine"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":18460,"name":"Jaiveek Shah","email":"jaiveek.shah@gmail.com","username":"JaiveekShah"},{"value":0,"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5558,"name":"Winson Chan","email":"wcchan@stackstorm.com","username":"winson-c-chan"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8358,"name":"zhangguoqing","email":"zhang.guoqing@99cloud.net","username":"zhangguoqing"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9432,"name":"Dmitri Zimine","email":"dz@stackstorm.com","username":"dzimine"},{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},{"_account_id":18460,"name":"Jaiveek Shah","email":"jaiveek.shah@gmail.com","username":"JaiveekShah"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-10-16 10:53:03.000000000","updated_by":{"_account_id":8358,"name":"zhangguoqing","email":"zhang.guoqing@99cloud.net","username":"zhangguoqing"},"reviewer":{"_account_id":8358,"name":"zhangguoqing","email":"zhang.guoqing@99cloud.net","username":"zhangguoqing"},"state":"REVIEWER"},{"updated":"2015-10-19 09:05:41.000000000","updated_by":{"_account_id":5558,"name":"Winson Chan","email":"wcchan@stackstorm.com","username":"winson-c-chan"},"reviewer":{"_account_id":5558,"name":"Winson Chan","email":"wcchan@stackstorm.com","username":"winson-c-chan"},"state":"REVIEWER"},{"updated":"2015-10-19 09:05:41.000000000","updated_by":{"_account_id":9432,"name":"Dmitri Zimine","email":"dz@stackstorm.com","username":"dzimine"},"reviewer":{"_account_id":9432,"name":"Dmitri Zimine","email":"dz@stackstorm.com","username":"dzimine"},"state":"REVIEWER"},{"updated":"2015-10-21 01:06:26.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2015-10-22 05:40:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2015-10-22 05:43:36.000000000","updated_by":{"_account_id":18460,"name":"Jaiveek Shah","email":"jaiveek.shah@gmail.com","username":"JaiveekShah"},"reviewer":{"_account_id":18460,"name":"Jaiveek Shah","email":"jaiveek.shah@gmail.com","username":"JaiveekShah"},"state":"REVIEWER"},{"updated":"2015-10-22 08:55:36.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2015-10-22 09:05:51.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"0b5026c3fbac6d79ad1e3f7bb75d613183c9c365","author":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"date":"2015-10-14 06:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2353805c95ae898160cf7f5a0e7cba294dbf0656","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-14 07:09:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-mistral-pep8 http://logs.openstack.org/87/234587/1/check/gate-mistral-pep8/95dd0bc/ : SUCCESS in 1m 50s\n- gate-mistral-docs http://docs-draft.openstack.org/87/234587/1/check/gate-mistral-docs/db261ce//doc/build/html/ : SUCCESS in 3m 43s\n- gate-mistral-python27 http://logs.openstack.org/87/234587/1/check/gate-mistral-python27/366dddc/ : SUCCESS in 6m 19s\n- gate-mistral-python34 http://logs.openstack.org/87/234587/1/check/gate-mistral-python34/fc4c706/ : FAILURE in 8m 32s (non-voting)\n- gate-rally-dsvm-mistral-task http://logs.openstack.org/87/234587/1/check/gate-rally-dsvm-mistral-task/120f092/ : SUCCESS in 28m 43s (non-voting)\n- gate-mistral-devstack-dsvm http://logs.openstack.org/87/234587/1/check/gate-mistral-devstack-dsvm/23638bf/ : SUCCESS in 42m 23s (non-voting)\n- gate-mistral-tox-unit-postgresql http://logs.openstack.org/87/234587/1/check/gate-mistral-tox-unit-postgresql/c841b52/ : FAILURE in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24666a0343a545cd428833889d520f4db59cdd0b","author":{"_account_id":8358,"name":"zhangguoqing","email":"zhang.guoqing@99cloud.net","username":"zhangguoqing"},"date":"2015-10-16 10:53:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2a0dd8c8cc14b15564603f3ef4b65d06d6c9ee67","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2015-10-19 10:18:02.000000000","message":"Patch Set 1: Code-Review-1\n\nHi, hardik, since this patch is part of python3 compatible effort, I wonder what\u0027s the exact error message in our python34 gate, if we don\u0027t use decode() method? It will be more useful for us to check the change is correct for the problem it intends to solve. You can provide such info included in your commit message, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"e29beaa3b30bee31ec320d362c761ceba2eef188","author":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"date":"2015-10-20 11:50:10.000000000","message":"Patch Set 2: Commit message was updated","accounts_in_message":[],"_revision_number":2},{"id":"044555af6bd841b1aad9e5306d63aeb9a774b745","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-20 12:30:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-mistral-pep8 http://logs.openstack.org/87/234587/2/check/gate-mistral-pep8/a5f3a6d/ : SUCCESS in 1m 58s\n- gate-mistral-docs http://docs-draft.openstack.org/87/234587/2/check/gate-mistral-docs/fb992e5//doc/build/html/ : SUCCESS in 2m 04s\n- gate-mistral-python27 http://logs.openstack.org/87/234587/2/check/gate-mistral-python27/b29c1e0/ : SUCCESS in 6m 20s\n- gate-mistral-python34 http://logs.openstack.org/87/234587/2/check/gate-mistral-python34/474227d/ : FAILURE in 8m 50s (non-voting)\n- gate-rally-dsvm-mistral-task http://logs.openstack.org/87/234587/2/check/gate-rally-dsvm-mistral-task/d8b018c/ : SUCCESS in 29m 29s (non-voting)\n- gate-mistral-devstack-dsvm http://logs.openstack.org/87/234587/2/check/gate-mistral-devstack-dsvm/435dd60/ : FAILURE in 36m 26s (non-voting)\n- gate-mistral-tox-unit-postgresql http://logs.openstack.org/87/234587/2/check/gate-mistral-tox-unit-postgresql/601aa24/ : FAILURE in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83c9cd90771afad24fc00496e9f66afba323f98c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2015-10-21 01:06:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2ddf886ce5c77b5d58927b9b1ac748392595ef01","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-21 01:43:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- gate-mistral-pep8 http://logs.openstack.org/87/234587/2/check/gate-mistral-pep8/1326afa/ : SUCCESS in 1m 30s\n- gate-mistral-docs http://docs-draft.openstack.org/87/234587/2/check/gate-mistral-docs/b83e1ec//doc/build/html/ : SUCCESS in 3m 01s\n- gate-mistral-python27 http://logs.openstack.org/87/234587/2/check/gate-mistral-python27/0cee64f/ : SUCCESS in 4m 59s\n- gate-mistral-python34 http://logs.openstack.org/87/234587/2/check/gate-mistral-python34/f95c25d/ : FAILURE in 8m 22s (non-voting)\n- gate-rally-dsvm-mistral-task http://logs.openstack.org/87/234587/2/check/gate-rally-dsvm-mistral-task/7c3d34f/ : SUCCESS in 23m 07s (non-voting)\n- gate-mistral-devstack-dsvm http://logs.openstack.org/87/234587/2/check/gate-mistral-devstack-dsvm/c705cec/ : SUCCESS in 37m 24s (non-voting)\n- gate-mistral-tox-unit-postgresql http://logs.openstack.org/87/234587/2/check/gate-mistral-tox-unit-postgresql/2cb0131/ : FAILURE in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c4745bb25cb4027dc1d6f6e5477f42e16d4e2c0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2015-10-21 09:39:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"44f5b98b668307497261205fa54126ee01c14f42","author":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"date":"2015-10-21 10:07:41.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"bfaed7fa55861a1e5824b53118627803316f3693","author":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"date":"2015-10-21 11:38:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b9ee2c2b585a49ba04d63a1a7443c71f77e89dfa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-21 12:15:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-mistral-pep8 http://logs.openstack.org/87/234587/3/check/gate-mistral-pep8/b998491/ : SUCCESS in 1m 43s\n- gate-mistral-docs http://docs-draft.openstack.org/87/234587/3/check/gate-mistral-docs/18cb271//doc/build/html/ : SUCCESS in 2m 06s\n- gate-mistral-python27 http://logs.openstack.org/87/234587/3/check/gate-mistral-python27/afe1f56/ : SUCCESS in 6m 02s\n- gate-mistral-python34 http://logs.openstack.org/87/234587/3/check/gate-mistral-python34/13835fe/ : FAILURE in 8m 11s (non-voting)\n- gate-rally-dsvm-mistral-task http://logs.openstack.org/87/234587/3/check/gate-rally-dsvm-mistral-task/128e3d2/ : FAILURE in 30m 50s (non-voting)\n- gate-mistral-devstack-dsvm http://logs.openstack.org/87/234587/3/check/gate-mistral-devstack-dsvm/19e09ff/ : SUCCESS in 35m 03s (non-voting)\n- gate-mistral-tox-unit-postgresql http://logs.openstack.org/87/234587/3/check/gate-mistral-tox-unit-postgresql/416ff6a/ : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4634fb78cc59e29c72810dba9bac961150339b7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2015-10-22 05:40:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"353e32eba4ecae7f2641caec056bf9aa50ca26f5","author":{"_account_id":18460,"name":"Jaiveek Shah","email":"jaiveek.shah@gmail.com","username":"JaiveekShah"},"date":"2015-10-22 05:43:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e0619b626b193790f56dd166fb1a71ac7b7a3754","author":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"date":"2015-10-22 08:55:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"51b6d854f94a9035904e4bda1b3d82817e83f1c2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-22 08:57:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"c1fe1a393cf1ea9c0bbc0313012318e64d5c1c68","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-22 09:05:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-mistral-docs http://docs-draft.openstack.org/87/234587/3/gate/gate-mistral-docs/df922a4//doc/build/html/ : SUCCESS in 1m 39s\n- gate-mistral-pep8 http://logs.openstack.org/87/234587/3/gate/gate-mistral-pep8/06d3f66/ : SUCCESS in 1m 46s\n- gate-mistral-python27 http://logs.openstack.org/87/234587/3/gate/gate-mistral-python27/f5e778d/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"dfcb666e251eff3ec8bb7947806b3aa41894458e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-22 09:05:53.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0ac5930bab19a7e33298a53f9ea63fe8e5dd1926","revisions":{"7e9cb6e6b99e099895bc72d8a8cc4e7e6f2e25de":{"kind":"REWORK","_number":1,"created":"2015-10-14 06:23:07.000000000","uploader":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"ref":"refs/changes/87/234587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/234587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/234587/1"}}},"commit":{"parents":[{"commit":"aebd8e706b9caaf201325bca07b15d02f5262a8b","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aebd8e706b9caaf201325bca07b15d02f5262a8b"}]}],"author":{"name":"hardik","email":"hardik.parekh@nectechnologies.in","date":"2015-09-26 22:27:35.000000000","tz":330},"committer":{"name":"hardik","email":"hardik.parekh@nectechnologies.in","date":"2015-09-26 23:56:15.000000000","tz":330},"subject":"Add decode() function for string comparision.","message":"Add decode() function for string comparision.\n\nA prefix of \u0027b\u0027 or \u0027B\u0027 is ignored in Python 2\nBut in py3 produce an instance of the bytes\ntype instead of the str type\n\nChange-Id: I226e8f8c6afbbc94c6348a17099f02b9cf7907a8\nPartially-Implements: blueprint mistral-py3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7e9cb6e6b99e099895bc72d8a8cc4e7e6f2e25de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7e9cb6e6b99e099895bc72d8a8cc4e7e6f2e25de"}]},"branch":"refs/heads/master"},"7b79fef0fbb96b068d921b95b6b1a13bc4f68dfb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-10-20 11:50:10.000000000","uploader":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"ref":"refs/changes/87/234587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/234587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/234587/2"}}},"commit":{"parents":[{"commit":"aebd8e706b9caaf201325bca07b15d02f5262a8b","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aebd8e706b9caaf201325bca07b15d02f5262a8b"}]}],"author":{"name":"hardik","email":"hardik.parekh@nectechnologies.in","date":"2015-09-26 22:27:35.000000000","tz":330},"committer":{"name":"hardik","email":"hardik.parekh@nectechnologies.in","date":"2015-10-20 11:50:10.000000000","tz":0},"subject":"Add decode() function for string comparison.","message":"Add decode() function for string comparison.\n\nA prefix of \u0027b\u0027 or \u0027B\u0027 is ignored in Python 2\nBut in py3 produce an instance of the bytes\ntype instead of the string type.\n\nDue to which it gives following kind of error.\ne.g.testtools.matchers._impl.MismatchError:\n\u0027unable to convert to int\u0027 not in \nb\u0027{\"faultcode\": \"Client\", \"debuginfo\": null, \n\"faultstring\": \"Invalid input for field/attribute limit. \nValue: \\\u00271.1\\\u0027. unable to convert to int\"}\u0027\n\nChange-Id: I226e8f8c6afbbc94c6348a17099f02b9cf7907a8\nPartially-Implements: blueprint mistral-py3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b79fef0fbb96b068d921b95b6b1a13bc4f68dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b79fef0fbb96b068d921b95b6b1a13bc4f68dfb"}]},"branch":"refs/heads/master"},"0ac5930bab19a7e33298a53f9ea63fe8e5dd1926":{"kind":"REWORK","_number":3,"created":"2015-10-21 11:38:44.000000000","uploader":{"_account_id":18238,"name":"Hardik Parekh","email":"hardik.parekh047@gmail.com","username":"hparekh"},"ref":"refs/changes/87/234587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/234587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/234587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/234587/3"}}},"commit":{"parents":[{"commit":"aebd8e706b9caaf201325bca07b15d02f5262a8b","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aebd8e706b9caaf201325bca07b15d02f5262a8b"}]}],"author":{"name":"hardik","email":"hardik.parekh@nectechnologies.in","date":"2015-09-26 22:27:35.000000000","tz":330},"committer":{"name":"hparekh","email":"hardik.parekh@nectechnologies.in","date":"2015-10-21 11:38:21.000000000","tz":330},"subject":"Add decode() function for string comparison.","message":"Add decode() function for string comparison.\n\nA prefix of \u0027b\u0027 or \u0027B\u0027 is ignored in Python 2\nBut in py3 produce an instance of the bytes\ntype instead of the string type.\n\nDue to which it gives following kind of error.\ne.g.testtools.matchers._impl.MismatchError:\n\u0027unable to convert to int\u0027 not in\nb\u0027{\"faultcode\": \"Client\", \"debuginfo\": null,\n\"faultstring\": \"Invalid input for field/attribute limit.\nValue: \\\u00271.1\\\u0027. unable to convert to int\"}\u0027\n\nChange-Id: I226e8f8c6afbbc94c6348a17099f02b9cf7907a8\nPartially-Implements: blueprint mistral-py3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0ac5930bab19a7e33298a53f9ea63fe8e5dd1926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0ac5930bab19a7e33298a53f9ea63fe8e5dd1926"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
