)]}'
{"id":"openstack%2Fmistral~424566","triplet_id":"openstack%2Fmistral~master~I3d8fe9a617d5e3f916a8c102d0408354064b766c","project":"openstack/mistral","branch":"master","topic":"invalid_jinja_regex","hashtags":[],"change_id":"I3d8fe9a617d5e3f916a8c102d0408354064b766c","subject":"Invalid jinja pattern regex corrected","status":"MERGED","created":"2017-01-24 10:28:13.000000000","updated":"2017-01-24 12:35:23.000000000","submitted":"2017-01-24 12:35:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd87208b1c511ff933eb2a0459dcfdceb6d909ab","_number":424566,"virtual_id_number":424566,"owner":{"_account_id":23271,"name":"Istvan Imre","email":"istvan.imre@nokia.com","username":"istvan.imre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2017-01-24 12:35:23.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":19206,"name":"Dawid Deja","email":"dawdej@gmail.com","username":"ddeja"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2017-01-24 11:37:56.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":19206,"name":"Dawid Deja","email":"dawdej@gmail.com","username":"ddeja"},{"value":2,"date":"2017-01-24 10:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":1,"date":"2017-01-24 11:37:56.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2017-01-24 10:42:44.000000000","_account_id":19206,"name":"Dawid Deja","email":"dawdej@gmail.com","username":"ddeja"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"date":"2017-01-24 10:42:52.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":19206,"name":"Dawid Deja","email":"dawdej@gmail.com","username":"ddeja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-24 10:42:44.000000000","updated_by":{"_account_id":19206,"name":"Dawid Deja","email":"dawdej@gmail.com","username":"ddeja"},"reviewer":{"_account_id":19206,"name":"Dawid Deja","email":"dawdej@gmail.com","username":"ddeja"},"state":"REVIEWER"},{"updated":"2017-01-24 10:42:52.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2017-01-24 10:50:32.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2017-01-24 11:37:56.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2017-01-24 12:35:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"8524feb7a95033b0066c08ec571327455bbb7c89","author":{"_account_id":23271,"name":"Istvan Imre","email":"istvan.imre@nokia.com","username":"istvan.imre"},"date":"2017-01-24 10:28:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eae762d643c5bd91c06fe0c2667a64717ea9b46c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2017-01-24 10:42:24.000000000","message":"Patch Set 1: Code-Review+1\n\nLet\u0027s wait for the tests. I wonder why we used question mark initially..","accounts_in_message":[],"_revision_number":1},{"id":"e874400e3c1c58a0063f8fa4b8426c47f41a4d46","author":{"_account_id":10127,"name":"Kirill Izotov","email":"enykeev@stackstorm.com","username":"enykeev"},"date":"2017-01-24 10:44:31.000000000","message":"Patch Set 1:\n\nI don\u0027t have any particular reason we may want to have `?` in this regexp. There is a good chance I\u0027ve just copied it from https://github.com/openstack/mistral/commit/b4a7ebeb0131bdf397bed8834b8341c3a26918be#diff-77f0dc3cce1911a102100602983cae8aR79 without giving it too much thought.\n\nWhat particular python environments regex like that may affect? It might be inefficient, but is it really invalid?","accounts_in_message":[],"_revision_number":1},{"id":"08ce65d3133da6d67e7a3a4e3c815c05d4f20b72","author":{"_account_id":23271,"name":"Istvan Imre","email":"istvan.imre@nokia.com","username":"istvan.imre"},"date":"2017-01-24 10:47:49.000000000","message":"Patch Set 1:\n\nThe following page also says that the given regex is invalid: http://pythex.org/\n\nAnd in our environment jinja does not work due to this invalid regex.","accounts_in_message":[],"_revision_number":1},{"id":"14ca777b9bad2084f85a77b20c271c94853ae132","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2017-01-24 10:50:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c94dda963b20e194b13f85825dd7765826bcb4c3","author":{"_account_id":10127,"name":"Kirill Izotov","email":"enykeev@stackstorm.com","username":"enykeev"},"date":"2017-01-24 10:58:21.000000000","message":"Patch Set 1:\n\nI went ahead and tested the regexp in several other validators from the top of the google search including http://rubular.com/r/c3I2kFaF7p, http://regexr.com/3f4o2, http://www.regexpal.com/?fam\u003d96726 and https://regex101.com/r/NGnPTq. None of them consider this regex being invalid.\n\nWhat I\u0027m trying to figure out here is what exactly is so special about the environment you\u0027re running this code in so we can avoid having similar problems in future.","accounts_in_message":[],"_revision_number":1},{"id":"e092b4a6c51eddc6af9c2d0ceff2839e7a98912c","author":{"_account_id":23271,"name":"Istvan Imre","email":"istvan.imre@nokia.com","username":"istvan.imre"},"date":"2017-01-24 11:08:40.000000000","message":"Patch Set 1:\n\nOur environment:\nRedHat linux with Python 2.7.5\n\nAnd the output of pip freeze:\nalembic\u003d\u003d0.8.9\namqp\u003d\u003d1.4.9\namqplib\u003d\u003d1.0.2\nansible\u003d\u003d2.1.0.0\nanyjson\u003d\u003d0.3.3\naodhclient\u003d\u003d0.7.0\nappdirs\u003d\u003d1.4.0\nBabel\u003d\u003d2.3.4\nbeautifulsoup4\u003d\u003d4.5.3\ncachetools\u003d\u003d2.0.0\ncertifi\u003d\u003d2016.9.26\ncffi\u003d\u003d1.9.1\ncliff\u003d\u003d2.4.0\ncmd2\u003d\u003d0.6.9\ncontextlib2\u003d\u003d0.5.4\ncroniter\u003d\u003d0.3.13\ncryptography\u003d\u003d1.7.1\ndebtcollector\u003d\u003d1.10.0\ndecorator\u003d\u003d4.0.10\ndogpile.cache\u003d\u003d0.6.2\nenum34\u003d\u003d1.1.6\neventlet\u003d\u003d0.19.0\nFabric\u003d\u003d1.13.1\nfasteners\u003d\u003d0.14.1\nfuncsigs\u003d\u003d1.0.2\nfunctools32\u003d\u003d3.2.3.post2\nfutures\u003d\u003d3.0.5\nfuturist\u003d\u003d0.20.0\ngreenlet\u003d\u003d0.4.10\nhttpretty\u003d\u003d0.8.14\nidna\u003d\u003d2.2\ninflection\u003d\u003d0.3.1\nipaddress\u003d\u003d1.0.18\niso8601\u003d\u003d0.1.11\nJinja2\u003d\u003d2.9.4\njsonpatch\u003d\u003d1.15\njsonpointer\u003d\u003d1.10\njsonschema\u003d\u003d2.5.1\nkazoo\u003d\u003d2.2.1\nkeystoneauth1\u003d\u003d2.18.0\nkeystonemiddleware\u003d\u003d4.11.0\nkombu\u003d\u003d3.0.37\nlogutils\u003d\u003d0.3.3\nMako\u003d\u003d1.0.6\nMarkupSafe\u003d\u003d0.23\nmistral\u003d\u003d3.0.0.0rc2.dev341\nmonotonic\u003d\u003d1.2\nmsgpack-python\u003d\u003d0.4.8\nMySQL-python\u003d\u003d1.2.5\nndg-httpsclient\u003d\u003d0.4.2\nnetaddr\u003d\u003d0.7.19\nnetifaces\u003d\u003d0.10.5\nnetworkx\u003d\u003d1.11\noauthlib\u003d\u003d2.0.1\nopenstacksdk\u003d\u003d0.9.10\nos-client-config\u003d\u003d1.24.0\nosc-lib\u003d\u003d1.3.0\noslo.concurrency\u003d\u003d3.16.0\noslo.config\u003d\u003d3.21.0\noslo.context\u003d\u003d2.11.0\noslo.db\u003d\u003d4.16.0\noslo.i18n\u003d\u003d3.12.0\noslo.log\u003d\u003d3.11.0\noslo.messaging\u003d\u003d5.14.0\noslo.middleware\u003d\u003d3.22.0\noslo.policy\u003d\u003d1.17.0\noslo.serialization\u003d\u003d2.15.0\noslo.service\u003d\u003d1.18.0\noslo.utils\u003d\u003d3.22.0\noslosphinx\u003d\u003d4.9.0\nosprofiler\u003d\u003d1.4.0\nparamiko\u003d\u003d2.1.1\nPaste\u003d\u003d2.0.3\nPasteDeploy\u003d\u003d1.5.2\npbr\u003d\u003d1.10.0\npecan\u003d\u003d1.1.2\npika\u003d\u003d0.10.0\npika-pool\u003d\u003d0.1.3\nply\u003d\u003d3.9\npositional\u003d\u003d1.1.1\nprettytable\u003d\u003d0.7.2\npyasn1\u003d\u003d0.1.9\npycadf\u003d\u003d2.4.0\npycparser\u003d\u003d2.17\npycrypto\u003d\u003d2.6.1\npyinotify\u003d\u003d0.9.6\npyOpenSSL\u003d\u003d16.2.0\npyparsing\u003d\u003d2.1.10\npython-barbicanclient\u003d\u003d4.1.0\npython-ceilometerclient\u003d\u003d2.7.0\npython-cinderclient\u003d\u003d1.10.0\npython-dateutil\u003d\u003d2.6.0\npython-designateclient\u003d\u003d2.3.0\npython-editor\u003d\u003d1.0.1\npython-glanceclient\u003d\u003d2.5.0\npython-heatclient\u003d\u003d1.7.0\npython-ironic-inspector-client\u003d\u003d1.9.0\npython-ironicclient\u003d\u003d1.8.0\npython-keystoneclient\u003d\u003d3.9.0\npython-magnumclient\u003d\u003d2.3.1\npython-mistralclient\u003d\u003d2.1.2.dev48\npython-muranoclient\u003d\u003d0.11.1\npython-neutronclient\u003d\u003d6.0.0\npython-novaclient\u003d\u003d6.0.0\npython-openstackclient\u003d\u003d3.7.0\npython-senlinclient\u003d\u003d1.1.0\npython-swiftclient\u003d\u003d3.2.0\npython-tackerclient\u003d\u003d0.7.0\npython-troveclient\u003d\u003d2.6.0\npython-zaqarclient\u003d\u003d1.2.0\npytz\u003d\u003d2016.10\nPyYAML\u003d\u003d3.12\nrepoze.lru\u003d\u003d0.6\nrequests\u003d\u003d2.12.5\nrequestsexceptions\u003d\u003d1.1.3\nrfc3986\u003d\u003d0.4.1\nRoutes\u003d\u003d2.3.1\nsimplegeneric\u003d\u003d0.8.1\nsimplejson\u003d\u003d3.10.0\nsingledispatch\u003d\u003d3.4.0.3\nsix\u003d\u003d1.10.0\nSQLAlchemy\u003d\u003d1.0.16\nsqlalchemy-migrate\u003d\u003d0.10.0\nsqlparse\u003d\u003d0.2.2\nstevedore\u003d\u003d1.20.0\nTempita\u003d\u003d0.5.2\ntenacity\u003d\u003d3.4.0\ntooz\u003d\u003d1.46.0\nunicodecsv\u003d\u003d0.14.1\nurllib3\u003d\u003d1.19.1\nvoluptuous\u003d\u003d0.9.3\nwaitress\u003d\u003d1.0.1\nwarlock\u003d\u003d1.2.0\nWebOb\u003d\u003d1.6.1\nWebTest\u003d\u003d2.0.24\nwrapt\u003d\u003d1.10.8\nWSME\u003d\u003d0.8.0\nyaql\u003d\u003d1.1.1\nzake\u003d\u003d0.2.2","accounts_in_message":[],"_revision_number":1},{"id":"cb7f4b7f1fbf86b9e397a2c3f1d8da4d4de16b39","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 11:24:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-mistral-docs-ubuntu-xenial http://docs-draft.openstack.org/66/424566/1/check/gate-mistral-docs-ubuntu-xenial/3e36785//doc/build/html/ : SUCCESS in 2m 43s\n- gate-mistral-pep8-ubuntu-xenial http://logs.openstack.org/66/424566/1/check/gate-mistral-pep8-ubuntu-xenial/f5aa572/ : SUCCESS in 3m 43s\n- gate-mistral-python27-ubuntu-xenial http://logs.openstack.org/66/424566/1/check/gate-mistral-python27-ubuntu-xenial/92692a4/ : SUCCESS in 6m 32s\n- gate-mistral-python35 http://logs.openstack.org/66/424566/1/check/gate-mistral-python35/7feecd4/ : SUCCESS in 5m 47s\n- mistral-coverage-ubuntu-xenial http://logs.openstack.org/66/424566/1/check/mistral-coverage-ubuntu-xenial/fe211b8/ : SUCCESS in 6m 06s (non-voting)\n- gate-rally-dsvm-mistral-task-ubuntu-xenial-nv http://logs.openstack.org/66/424566/1/check/gate-rally-dsvm-mistral-task-ubuntu-xenial-nv/1db28aa/ : SUCCESS in 43m 35s (non-voting)\n- gate-mistral-devstack-dsvm-ubuntu-xenial-nv http://logs.openstack.org/66/424566/1/check/gate-mistral-devstack-dsvm-ubuntu-xenial-nv/4df9c01/ : SUCCESS in 24m 22s (non-voting)\n- gate-mistral-tox-db-unit-mysql-ubuntu-xenial http://logs.openstack.org/66/424566/1/check/gate-mistral-tox-db-unit-mysql-ubuntu-xenial/8a5f23a/ : SUCCESS in 21m 06s (non-voting)\n- gate-mistral-tox-db-unit-postgresql-ubuntu-xenial http://logs.openstack.org/66/424566/1/check/gate-mistral-tox-db-unit-postgresql-ubuntu-xenial/1d66323/ : SUCCESS in 17m 00s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/66/424566/1/check/gate-tripleo-ci-centos-7-nonha-multinode/e6f04a2/ : SUCCESS in 54m 27s\n- gate-mistral-devstack-dsvm-kombu-nv http://logs.openstack.org/66/424566/1/check/gate-mistral-devstack-dsvm-kombu-nv/e24ed30/ : FAILURE in 41m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d383d17955d6bbed13c54375c3f47301cc02d4e","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2017-01-24 11:37:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis seems to be a limitation of python 2.7.5, it works fine in a more recent 2.7.x.","accounts_in_message":[],"_revision_number":1},{"id":"090d462cbe6159809b06b83b1f4da4acebe6342f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 11:38:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"64f443fd4881f8edb9268572abc7c6ceedf60e2c","date":"2017-01-24 12:35:23.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"9c67331243b14c44bac5d71b2ace5cbed06cb460","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 12:35:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-mistral-docs-ubuntu-xenial http://docs-draft.openstack.org/66/424566/1/gate/gate-mistral-docs-ubuntu-xenial/9e44e53//doc/build/html/ : SUCCESS in 6m 27s\n- gate-mistral-pep8-ubuntu-xenial http://logs.openstack.org/66/424566/1/gate/gate-mistral-pep8-ubuntu-xenial/62a8a6a/ : SUCCESS in 3m 08s\n- gate-mistral-python27-ubuntu-xenial http://logs.openstack.org/66/424566/1/gate/gate-mistral-python27-ubuntu-xenial/3b3e5d2/ : SUCCESS in 6m 25s\n- gate-mistral-python35 http://logs.openstack.org/66/424566/1/gate/gate-mistral-python35/2153de7/ : SUCCESS in 5m 55s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/66/424566/1/gate/gate-tripleo-ci-centos-7-nonha-multinode/9450979/ : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5267e1eb1c4e7f9c619cc24a9a5f9e23abcca2a8","revisions":{"5267e1eb1c4e7f9c619cc24a9a5f9e23abcca2a8":{"kind":"REWORK","_number":1,"created":"2017-01-24 10:28:13.000000000","uploader":{"_account_id":23271,"name":"Istvan Imre","email":"istvan.imre@nokia.com","username":"istvan.imre"},"ref":"refs/changes/66/424566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/66/424566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/424566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/424566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/66/424566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/66/424566/1"}}},"commit":{"parents":[{"commit":"93ecd38dce4864d7299e423206e0edf62a54c038","subject":"Merge \"Allow hyphens in Workflow and ad-hoc action names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/93ecd38dce4864d7299e423206e0edf62a54c038"}]}],"author":{"name":"Istvan Imre","email":"istvan.imre@nokia.com","date":"2017-01-24 10:26:08.000000000","tz":60},"committer":{"name":"Istvan Imre","email":"istvan.imre@nokia.com","date":"2017-01-24 10:26:08.000000000","tz":60},"subject":"Invalid jinja pattern regex corrected","message":"Invalid jinja pattern regex corrected\n\nDue to this invalid regex it is not possible to run workflows\nwith Jinja2 expressions in some python2 environment.\n\nChange-Id: I3d8fe9a617d5e3f916a8c102d0408354064b766c\nCloses-bug: #1658958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5267e1eb1c4e7f9c619cc24a9a5f9e23abcca2a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5267e1eb1c4e7f9c619cc24a9a5f9e23abcca2a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
