)]}'
{"id":"openstack%2Fmistral~557487","triplet_id":"openstack%2Fmistral~master~I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b","project":"openstack/mistral","branch":"master","topic":"557487","hashtags":[],"change_id":"I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b","subject":"Use eventlet-aware threading events","status":"MERGED","created":"2018-03-28 19:07:15.000000000","updated":"2019-06-18 06:35:20.000000000","submitted":"2019-06-18 06:35:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"557487-1560839720205-d3b36feb","meta_rev_id":"92d0858581ca97da5fb95aaa9de4cb24a21c8720","_number":557487,"virtual_id_number":557487,"owner":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-06-18 06:35:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2019-06-17 14:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-06-18 04:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-06-18 04:54:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-24 19:03:35.000000000","updated_by":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"reviewer":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"state":"REVIEWER"},{"updated":"2018-09-25 22:49:26.000000000","updated_by":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"reviewer":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"state":"REVIEWER"},{"updated":"2018-10-08 12:02:44.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2018-10-14 16:24:33.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2018-11-19 08:46:48.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2018-11-27 08:24:50.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-06-09 10:15:29.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-06-09 15:38:52.000000000","updated_by":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"reviewer":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"state":"REVIEWER"},{"updated":"2019-06-17 14:59:29.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-06-18 04:54:57.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-06-18 06:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a92c2a39b29a151769f6baf4ed761d11e37440e","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2018-03-28 19:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8f0f2ead1622ec3d58a946e5efaf246506d89d2","author":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"date":"2018-03-28 19:20:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b6ec61127e4bdf3050b03f64814e3f449db94b7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 21:08:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/1/check/openstack-tox-pep8/b8ebacf/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/1/check/openstack-tox-py27/11389a8/ : FAILURE in 8m 53s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/1/check/openstack-tox-py35/dd82916/ : FAILURE in 8m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/557487/1/check/build-openstack-sphinx-docs/76a402d/html/ : SUCCESS in 5m 42s\n- openstack-tox-cover http://logs.openstack.org/87/557487/1/check/openstack-tox-cover/ff38a7e/ : FAILURE in 11m 53s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/1/check/mistral-devstack/bd4e62e/ : SUCCESS in 35m 25s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/1/check/mistral-devstack-non-apache/489ae45/ : SUCCESS in 37m 38s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/1/check/mistral-devstack-kombu/a92d0e4/ : SUCCESS in 41m 14s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/1/check/mistral-tox-unit-mysql/def9236/ : FAILURE in 25m 54s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/1/check/mistral-tox-unit-postgresql/9969ecd/ : FAILURE in 25m 22s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/35d8bfd/ : SUCCESS in 1h 57m 55s\n- mistral-rally-task http://logs.openstack.org/87/557487/1/check/mistral-rally-task/9c293ec/ : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ff71cefa9dfc0838deb04774d44b2ff867cc1605","author":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"date":"2018-03-29 12:18:44.000000000","message":"Patch Set 1: Code-Review-1\n\nThe same test failed in 4 of the jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7352ab94cd2417e2bd1bf145f45232b38d40111a","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2018-04-04 18:58:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c17b496b6e72ce1e0ae01ac42c4d75a575980c18","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2018-04-04 18:59:43.000000000","message":"Patch Set 2:\n\n\u003e The same test failed in 4 of the jobs.\n\nAdded Depends-On for a needed oslo.utils change.  Tests pass for me locally when I include it.","accounts_in_message":[],"_revision_number":2},{"id":"d0733c137c859b06974a6d7bc818946990c258e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-04 20:14:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/2/check/openstack-tox-pep8/9852fc1/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/2/check/openstack-tox-py27/e7050cd/ : FAILURE in 7m 40s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/2/check/openstack-tox-py35/bbf696b/ : FAILURE in 8m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/557487/2/check/build-openstack-sphinx-docs/8d5bb86/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover http://logs.openstack.org/87/557487/2/check/openstack-tox-cover/197a59b/ : FAILURE in 11m 53s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/2/check/mistral-devstack/60fce0e/ : SUCCESS in 48m 20s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/2/check/mistral-devstack-non-apache/d4439ef/ : SUCCESS in 34m 00s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/2/check/mistral-devstack-kombu/fe83ee8/ : SUCCESS in 34m 12s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/2/check/mistral-tox-unit-mysql/c27fe3f/ : FAILURE in 25m 09s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/2/check/mistral-tox-unit-postgresql/d469df4/ : FAILURE in 26m 18s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/a555243/ : FAILURE in 1h 10m 11s\n- mistral-rally-task http://logs.openstack.org/87/557487/2/check/mistral-rally-task/0123467/ : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":2},{"id":"1f55387a90c30969ac35f8c6c02f9a7e607876bd","author":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"date":"2018-09-24 19:03:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9b68e8e72ea0f669642476a0034dab09efc7632d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-25 03:50:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/2/check/openstack-tox-pep8/c25b3d4/ : SUCCESS in 5m 51s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/2/check/openstack-tox-py27/4ea2800/ : SUCCESS in 9m 04s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/2/check/openstack-tox-py35/fcc878b/ : SUCCESS in 8m 52s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/2/check/openstack-tox-py36/4ea0770/ : SUCCESS in 9m 34s\n- openstack-tox-docs http://logs.openstack.org/87/557487/2/check/openstack-tox-docs/3965257/html/ : SUCCESS in 5m 07s\n- openstack-tox-cover http://logs.openstack.org/87/557487/2/check/openstack-tox-cover/c392677/cover/ : SUCCESS in 10m 09s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/2/check/mistral-devstack/c38f27a/ : SUCCESS in 44m 53s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/2/check/mistral-devstack-non-apache/5172e60/ : SUCCESS in 36m 53s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/2/check/mistral-devstack-kombu/1f38c4e/ : SUCCESS in 47m 17s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/2/check/mistral-tox-unit-mysql/8b1b5ce/ : SUCCESS in 30m 01s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/2/check/mistral-tox-unit-postgresql/1fc7543/ : SUCCESS in 28m 04s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/e8fe2d7/ : SUCCESS in 2h 32m 20s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/2/check/tripleo-ci-centos-7-undercloud-containers/0731384/ : SUCCESS in 1h 35m 07s\n- mistral-rally-task http://logs.openstack.org/87/557487/2/check/mistral-rally-task/c50d3bf/ : FAILURE in 1h 59m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/2/check/openstack-tox-lower-constraints/0c4e6a2/ : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"cb9cfd9b14c37bfd5619d008f006fa11e5360455","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-09-25 08:38:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"13864db435ee3ea1e925ff451905c98bc22f9321","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-25 13:21:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/2/check/openstack-tox-pep8/d33e7d9/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/2/check/openstack-tox-py27/2c511f2/ : SUCCESS in 10m 20s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/2/check/openstack-tox-py35/3bcdf32/ : SUCCESS in 9m 37s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/2/check/openstack-tox-py36/c372ef7/ : SUCCESS in 9m 15s\n- openstack-tox-docs http://logs.openstack.org/87/557487/2/check/openstack-tox-docs/676f1bc/html/ : SUCCESS in 5m 08s\n- openstack-tox-cover http://logs.openstack.org/87/557487/2/check/openstack-tox-cover/0f004b1/cover/ : SUCCESS in 9m 42s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/2/check/mistral-devstack/b1f99dc/ : FAILURE in 43m 31s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/2/check/mistral-devstack-non-apache/6f693e0/ : SUCCESS in 45m 46s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/2/check/mistral-devstack-kombu/7e56b78/ : FAILURE in 44m 09s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/2/check/mistral-tox-unit-mysql/3eba913/ : SUCCESS in 30m 15s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/2/check/mistral-tox-unit-postgresql/a548724/ : SUCCESS in 27m 55s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/2cdb7df/ : SUCCESS in 3h 31m 30s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/2/check/tripleo-ci-centos-7-undercloud-containers/e297971/ : FAILURE in 1h 28m 29s\n- mistral-rally-task http://logs.openstack.org/87/557487/2/check/mistral-rally-task/420b7e9/ : SUCCESS in 1h 32m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/2/check/openstack-tox-lower-constraints/a3a6e33/ : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f3bf70509dbc70cfdbdd361d8396e5704ccbecb3","author":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"date":"2018-09-25 14:15:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"663324bc541d6968efd6e28b459b3043da2e0d1d","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2018-09-25 14:21:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63195a6f6c92529b127c25d2aa5dd1c45109741f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-25 20:05:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/3/check/openstack-tox-pep8/c1e87bf/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/3/check/openstack-tox-py27/39ce22c/ : SUCCESS in 8m 38s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/3/check/openstack-tox-py35/afb3c79/ : SUCCESS in 9m 36s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/3/check/openstack-tox-py36/701a51d/ : SUCCESS in 9m 19s\n- openstack-tox-docs http://logs.openstack.org/87/557487/3/check/openstack-tox-docs/5c3ee3d/html/ : SUCCESS in 6m 13s\n- openstack-tox-cover http://logs.openstack.org/87/557487/3/check/openstack-tox-cover/914d3c8/cover/ : SUCCESS in 10m 36s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/3/check/mistral-devstack/bb1c6e2/ : FAILURE in 39m 33s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/3/check/mistral-devstack-non-apache/cbf7e24/ : SUCCESS in 38m 26s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/3/check/mistral-devstack-kombu/0bfe7fd/ : FAILURE in 38m 29s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-mysql/41553a9/ : SUCCESS in 30m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-postgresql/580ef58/ : SUCCESS in 27m 54s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/cd47f02/ : SUCCESS in 3h 14m 53s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-undercloud-containers/73f5f20/ : FAILURE in 1h 45m 40s\n- mistral-rally-task http://logs.openstack.org/87/557487/3/check/mistral-rally-task/575b3fe/ : FAILURE in 1h 50m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/3/check/openstack-tox-lower-constraints/443152d/ : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":3},{"id":"1540c313c49b9f77da69bb5efbfb045e9b67544f","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-09-25 22:49:26.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"28b5204ee5dfa1a0df9cd372372f97e277377113","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-26 05:54:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/3/check/openstack-tox-pep8/d4bef31/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/3/check/openstack-tox-py27/dbf2ffd/ : SUCCESS in 9m 20s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/3/check/openstack-tox-py35/3fe2449/ : SUCCESS in 9m 48s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/3/check/openstack-tox-py36/886c567/ : SUCCESS in 9m 21s\n- openstack-tox-docs http://logs.openstack.org/87/557487/3/check/openstack-tox-docs/a96068d/html/ : SUCCESS in 4m 43s\n- openstack-tox-cover http://logs.openstack.org/87/557487/3/check/openstack-tox-cover/057524b/cover/ : SUCCESS in 10m 14s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/3/check/mistral-devstack/3d7db94/ : SUCCESS in 47m 06s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/3/check/mistral-devstack-non-apache/4698c00/ : SUCCESS in 46m 55s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/3/check/mistral-devstack-kombu/8cb98ed/ : SUCCESS in 35m 42s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-mysql/8c73176/ : FAILURE in 30m 13s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-postgresql/5abd01c/ : SUCCESS in 29m 21s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/67e8bc6/ : SUCCESS in 2h 49m 43s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-undercloud-containers/3eda3ec/ : SUCCESS in 1h 43m 24s\n- mistral-rally-task http://logs.openstack.org/87/557487/3/check/mistral-rally-task/73012d3/ : FAILURE in 1h 56m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/3/check/openstack-tox-lower-constraints/38f4e7d/ : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":3},{"id":"edc349f53e128a31df87f9016b56fbda7ffaf92a","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2018-09-27 14:06:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1e5cb49daf7688b1b5cb9706e582002bce3a0882","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 21:14:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/3/check/openstack-tox-pep8/3e15c94/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/3/check/openstack-tox-py27/fb08a85/ : SUCCESS in 9m 09s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/3/check/openstack-tox-py35/ed67acb/ : SUCCESS in 9m 11s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/3/check/openstack-tox-py36/26805e5/ : SUCCESS in 9m 22s\n- openstack-tox-docs http://logs.openstack.org/87/557487/3/check/openstack-tox-docs/8df4b7d/html/ : SUCCESS in 6m 55s\n- openstack-tox-cover http://logs.openstack.org/87/557487/3/check/openstack-tox-cover/905b0c5/cover/ : SUCCESS in 8m 52s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/3/check/mistral-devstack/96f04b1/ : SUCCESS in 41m 42s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/3/check/mistral-devstack-non-apache/c35c5d9/ : SUCCESS in 43m 44s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/3/check/mistral-devstack-kombu/51902c4/ : SUCCESS in 35m 39s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-mysql/e90939d/ : SUCCESS in 30m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-postgresql/15bb2f0/ : SUCCESS in 28m 13s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/9cf5e9a/ : SUCCESS in 2h 27m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-undercloud-containers/981b835/ : SUCCESS in 1h 40m 22s\n- mistral-rally-task http://logs.openstack.org/87/557487/3/check/mistral-rally-task/246698a/ : FAILURE in 1h 36m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/3/check/openstack-tox-lower-constraints/b74e0fa/ : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":3},{"id":"671f97ad8d1d225a5f99b130c739366d9e6bf5e4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-02 03:15:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c48e03ebe438f502ca4ae46f5c808cc5427a2110","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-02 05:23:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/3/check/openstack-tox-pep8/99ad188/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/3/check/openstack-tox-py27/4aa969a/ : SUCCESS in 9m 43s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/3/check/openstack-tox-py35/97d2f97/ : SUCCESS in 9m 31s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/3/check/openstack-tox-py36/6554aba/ : SUCCESS in 9m 26s\n- openstack-tox-docs http://logs.openstack.org/87/557487/3/check/openstack-tox-docs/74f59b2/html/ : SUCCESS in 5m 05s\n- openstack-tox-cover http://logs.openstack.org/87/557487/3/check/openstack-tox-cover/6f8886c/cover/ : SUCCESS in 10m 50s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/3/check/mistral-devstack/c77e799/ : SUCCESS in 43m 10s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/3/check/mistral-devstack-non-apache/84f942e/ : SUCCESS in 44m 51s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/3/check/mistral-devstack-kombu/3b98b1c/ : SUCCESS in 44m 56s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-mysql/538ea73/ : SUCCESS in 31m 45s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-postgresql/3170cf2/ : SUCCESS in 28m 09s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/23290f6/ : FAILURE in 1h 59m 58s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-undercloud-containers/741bdd8/ : SUCCESS in 1h 41m 36s\n- mistral-rally-task http://logs.openstack.org/87/557487/3/check/mistral-rally-task/3e2ce9f/ : POST_FAILURE in 2h 06m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/3/check/openstack-tox-lower-constraints/c94570b/ : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":3},{"id":"59b77053fba9550ed243349068e2c6d494229795","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-02 06:17:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5d3df6fdc359bb3cabfbbed7ad6c797b33eecd3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-02 08:51:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/3/check/openstack-tox-pep8/0e19e9a/ : SUCCESS in 6m 11s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/3/check/openstack-tox-py27/3cae930/ : SUCCESS in 10m 29s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/3/check/openstack-tox-py35/24e1c0f/ : SUCCESS in 10m 42s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/3/check/openstack-tox-py36/0606302/ : SUCCESS in 10m 14s\n- openstack-tox-docs http://logs.openstack.org/87/557487/3/check/openstack-tox-docs/9e7ae6e/html/ : SUCCESS in 8m 23s\n- openstack-tox-cover http://logs.openstack.org/87/557487/3/check/openstack-tox-cover/972cc3e/cover/ : SUCCESS in 11m 01s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/3/check/mistral-devstack/59db810/ : SUCCESS in 40m 55s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/3/check/mistral-devstack-non-apache/203b69e/ : SUCCESS in 49m 02s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/3/check/mistral-devstack-kombu/ad90299/ : SUCCESS in 47m 52s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-mysql/b6aa306/ : SUCCESS in 31m 33s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/3/check/mistral-tox-unit-postgresql/48dcc58/ : SUCCESS in 30m 30s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/750ada4/ : SUCCESS in 2h 24m 19s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/3/check/tripleo-ci-centos-7-undercloud-containers/524e733/ : SUCCESS in 1h 24m 39s\n- mistral-rally-task http://logs.openstack.org/87/557487/3/check/mistral-rally-task/4245c9e/ : POST_FAILURE in 1h 44m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/3/check/openstack-tox-lower-constraints/965fdb7/ : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":3},{"id":"871f5ea6d28afd064a76374a789100063c815c3a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-02 09:31:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e211a321661363b8c68305117666a2e101d987d9","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-10-02 11:13:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"12bd479e020762823ff51a3b79d84e157735c2cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-02 14:54:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/4/check/openstack-tox-pep8/1a6a22a/ : SUCCESS in 8m 55s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/4/check/openstack-tox-py27/530b3ed/ : SUCCESS in 9m 46s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/4/check/openstack-tox-py35/5c08255/ : SUCCESS in 8m 52s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/4/check/openstack-tox-py36/137ad3c/ : SUCCESS in 8m 48s\n- openstack-tox-docs http://logs.openstack.org/87/557487/4/check/openstack-tox-docs/c89a0dd/html/ : SUCCESS in 7m 36s\n- openstack-tox-cover http://logs.openstack.org/87/557487/4/check/openstack-tox-cover/8998a0a/cover/ : SUCCESS in 8m 53s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/4/check/mistral-devstack/e5f7ac5/ : SUCCESS in 46m 45s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/4/check/mistral-devstack-non-apache/1201a91/ : SUCCESS in 40m 13s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/4/check/mistral-devstack-kombu/7c8f0f3/ : SUCCESS in 44m 16s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/4/check/mistral-tox-unit-mysql/6cae7c4/ : SUCCESS in 31m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/4/check/mistral-tox-unit-postgresql/ef63c00/ : SUCCESS in 28m 21s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/4/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/fa5f1c7/ : TIMED_OUT in 3h 32m 23s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/4/check/tripleo-ci-centos-7-undercloud-containers/a62268d/ : SUCCESS in 1h 27m 19s\n- mistral-rally-task http://logs.openstack.org/87/557487/4/check/mistral-rally-task/ec14377/ : SUCCESS in 1h 38m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/4/check/openstack-tox-lower-constraints/fe1b5fb/ : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2710227ae8a5a446d2674de90f1c5498d708bf36","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-08 05:47:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f389ec3d84a9b67ee3d1be22489ec4552779e9f5","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-08 05:47:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e94957382f0c1c7583810844633cc16e0e825727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 08:41:51.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/4/check/openstack-tox-pep8/aeb948f/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/4/check/openstack-tox-py27/2551e4b/ : SUCCESS in 9m 43s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/4/check/openstack-tox-py35/ee8342a/ : SUCCESS in 10m 27s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/4/check/openstack-tox-py36/1ff7a32/ : SUCCESS in 9m 24s\n- openstack-tox-docs http://logs.openstack.org/87/557487/4/check/openstack-tox-docs/3bef781/html/ : SUCCESS in 5m 54s\n- openstack-tox-cover http://logs.openstack.org/87/557487/4/check/openstack-tox-cover/b595b6a/cover/ : SUCCESS in 9m 48s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/4/check/mistral-devstack/f7f7d19/ : SUCCESS in 45m 58s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/4/check/mistral-devstack-non-apache/009577c/ : SUCCESS in 41m 52s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/4/check/mistral-devstack-kombu/ae159e7/ : SUCCESS in 44m 50s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/4/check/mistral-tox-unit-mysql/e28ee5a/ : SUCCESS in 30m 42s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/4/check/mistral-tox-unit-postgresql/70f0d13/ : SUCCESS in 29m 19s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/4/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/7f6f728/ : SUCCESS in 2h 50m 30s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/4/check/tripleo-ci-centos-7-undercloud-containers/b28f1de/ : SUCCESS in 1h 41m 01s\n- mistral-rally-task http://logs.openstack.org/87/557487/4/check/mistral-rally-task/1e7306b/ : FAILURE in 1h 53m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/4/check/openstack-tox-lower-constraints/afb10bd/ : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":4},{"id":"f9910d8094a01ec95d92263bcfd9baa8b44d12cd","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-10-08 12:01:40.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c59aafaaaea92f98c78c6489e58d8ffb16a8a3dd","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-10-08 12:02:39.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"80e19656165fe1e3f7c2272c8395f4a2d0fb0afe","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-10-08 12:02:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nRally has passed previously. Lets see if we can get this through.","accounts_in_message":[],"_revision_number":5},{"id":"29295202a5497f4342b33df342b496ed40f4bba8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 15:01:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/5/check/openstack-tox-pep8/f9c26fc/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/5/check/openstack-tox-py27/928b1df/ : SUCCESS in 9m 57s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/5/check/openstack-tox-py35/f7f22e0/ : SUCCESS in 9m 53s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/5/check/openstack-tox-py36/46356a4/ : SUCCESS in 9m 48s\n- openstack-tox-docs http://logs.openstack.org/87/557487/5/check/openstack-tox-docs/03c4442/html/ : SUCCESS in 4m 56s\n- openstack-tox-cover http://logs.openstack.org/87/557487/5/check/openstack-tox-cover/4e5062f/cover/ : SUCCESS in 11m 06s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/5/check/mistral-devstack/d03a47b/ : SUCCESS in 43m 33s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/5/check/mistral-devstack-non-apache/13248ee/ : SUCCESS in 58m 33s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/5/check/mistral-devstack-kombu/f060adb/ : SUCCESS in 36m 25s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/5/check/mistral-tox-unit-mysql/f3fa933/ : SUCCESS in 30m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/5/check/mistral-tox-unit-postgresql/b2369c3/ : SUCCESS in 28m 08s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/5/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/4613cff/ : SUCCESS in 2h 38m 46s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/5/check/tripleo-ci-centos-7-undercloud-containers/b4b9546/ : SUCCESS in 1h 37m 33s\n- mistral-rally-task http://logs.openstack.org/87/557487/5/check/mistral-rally-task/ad05828/ : SUCCESS in 1h 40m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/5/check/openstack-tox-lower-constraints/b36823f/ : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":5},{"id":"10ab7b31cbc32790b47b52aa4d4610d9a8c3b1e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 15:01:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"28c32b4d0c8be28f1ee48d4e3595a32fb850f4d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-08 17:40:56.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/5/gate/openstack-tox-pep8/551b509/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/5/gate/openstack-tox-py27/9686d97/ : SUCCESS in 10m 26s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/5/gate/openstack-tox-py35/6f81d53/ : SUCCESS in 9m 41s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/5/gate/openstack-tox-py36/1bd4baf/ : SUCCESS in 10m 06s\n- openstack-tox-docs http://logs.openstack.org/87/557487/5/gate/openstack-tox-docs/833080d/html/ : SUCCESS in 5m 18s\n- mistral-devstack http://logs.openstack.org/87/557487/5/gate/mistral-devstack/b29210a/ : SUCCESS in 40m 08s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/5/gate/mistral-devstack-non-apache/f4cc772/ : SUCCESS in 41m 17s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/5/gate/mistral-tox-unit-mysql/956130a/ : SUCCESS in 32m 13s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/5/gate/mistral-tox-unit-postgresql/1f56785/ : SUCCESS in 28m 37s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/5/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/a828cb3/ : SUCCESS in 2h 33m 16s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/5/gate/mistral-devstack-kombu/33dffa8/ : SUCCESS in 47m 31s\n- mistral-rally-task http://logs.openstack.org/87/557487/5/gate/mistral-rally-task/d4d929b/ : SUCCESS in 1h 36m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/5/gate/openstack-tox-lower-constraints/b3c5da8/ : FAILURE in 11m 02s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/5/gate/tripleo-ci-centos-7-undercloud-containers/1454eef/ : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":5},{"id":"4ca5d7285e8489747415d0fc155c257f0ad8c6f9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-09 04:04:11.000000000","message":"Patch Set 5: Code-Review-1\n\nCan you please investigate the gate failure? It seems to be related with the change.","accounts_in_message":[],"_revision_number":5},{"id":"a724f8fa68c8860cd59215462cd4f502a275fbcd","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-09 05:40:01.000000000","message":"Patch Set 5: -Code-Review\n\nBTW, this kombu driver is not production ready. I tried to use it and it did not provide the robustness required in production. There are race conditions that can make it unstable under certain conditions. oslo.messaging may be a little slower but better takes on errors in the infrastructure.","accounts_in_message":[],"_revision_number":5},{"id":"d62f879a26e8cda32409dc84d46e43beeebef0ee","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-09 06:13:46.000000000","message":"Patch Set 5:\n\nYes, right. We hit these issues too and I tell everyone about it.","accounts_in_message":[],"_revision_number":5},{"id":"3fe4ab70da880a37ff22a3f632632f04121c704a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-14 10:28:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"88f5e8b5969202c657ea356a6fa56af12579f994","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-14 13:34:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/6/check/openstack-tox-pep8/334cf50/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/6/check/openstack-tox-py27/8219cd5/ : SUCCESS in 8m 57s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/6/check/openstack-tox-py35/9954c9e/ : SUCCESS in 8m 30s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/6/check/openstack-tox-py36/f0da950/ : SUCCESS in 8m 50s\n- openstack-tox-docs http://logs.openstack.org/87/557487/6/check/openstack-tox-docs/e19b9f6/html/ : SUCCESS in 4m 14s\n- openstack-tox-cover http://logs.openstack.org/87/557487/6/check/openstack-tox-cover/a5c1695/cover/ : SUCCESS in 8m 36s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/6/check/mistral-devstack/3c7a8a1/ : SUCCESS in 40m 11s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/6/check/mistral-devstack-non-apache/e12a0c6/ : SUCCESS in 59m 12s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/6/check/mistral-devstack-kombu/74d3e41/ : SUCCESS in 42m 57s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-mysql/ff093f5/ : SUCCESS in 29m 13s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-postgresql/20021f4/ : SUCCESS in 28m 33s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/a6bfe15/ : SUCCESS in 3h 00m 48s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-undercloud-containers/0c1bef2/ : SUCCESS in 1h 41m 00s\n- mistral-rally-task http://logs.openstack.org/87/557487/6/check/mistral-rally-task/6641fd2/ : SUCCESS in 1h 32m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/6/check/openstack-tox-lower-constraints/c00463a/ : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":6},{"id":"5054ae0cf013a50065aa967214a8373962a27d59","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-10-14 16:24:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"24e7866779429f36b2a8a3cc717603189e25d543","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2018-10-16 04:52:27.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ac6d421ab62fb6c225286957d3fd655d4862e617","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-16 04:52:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1e43737ce57d8c271003ba7f1769236b23c8e55f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-16 08:31:00.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/6/gate/openstack-tox-pep8/e0268b6/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py27/c899d0f/ : SUCCESS in 9m 19s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py35/5fed751/ : SUCCESS in 9m 40s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py36/814d96a/ : SUCCESS in 9m 24s\n- openstack-tox-docs http://logs.openstack.org/87/557487/6/gate/openstack-tox-docs/e321a73/html/ : SUCCESS in 4m 15s\n- mistral-devstack http://logs.openstack.org/87/557487/6/gate/mistral-devstack/cfc6652/ : SUCCESS in 44m 39s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/6/gate/mistral-devstack-non-apache/82b44b9/ : SUCCESS in 37m 53s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/6/gate/mistral-tox-unit-mysql/7f39791/ : SUCCESS in 31m 02s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/6/gate/mistral-tox-unit-postgresql/83b45aa/ : SUCCESS in 29m 44s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/6/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/060f09e/ : TIMED_OUT in 3h 35m 57s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/6/gate/mistral-devstack-kombu/3ec13e5/ : SUCCESS in 49m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/6/gate/openstack-tox-lower-constraints/c648689/ : FAILURE in 12m 05s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/6/gate/tripleo-ci-centos-7-undercloud-containers/a6098ea/ : SUCCESS in 1h 42m 51s","accounts_in_message":[],"_revision_number":6},{"id":"5b60d3587f8a9a7548398ae06662d383ad2100e7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-10-17 06:34:23.000000000","message":"Patch Set 6: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a3562509365cceb9dee730a3e87736a65eb863ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-17 10:01:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/6/check/openstack-tox-pep8/54293b7/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/6/check/openstack-tox-py27/8518e91/ : SUCCESS in 8m 45s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/6/check/openstack-tox-py35/c3bbd6d/ : SUCCESS in 9m 28s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/6/check/openstack-tox-py36/6117f06/ : SUCCESS in 8m 45s\n- openstack-tox-docs http://logs.openstack.org/87/557487/6/check/openstack-tox-docs/ae8d311/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/87/557487/6/check/openstack-tox-cover/6ca89c4/cover/ : SUCCESS in 9m 08s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/6/check/mistral-devstack/0e403e8/ : SUCCESS in 49m 12s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/6/check/mistral-devstack-non-apache/c65368e/ : SUCCESS in 35m 39s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/6/check/mistral-devstack-kombu/d30b7e9/ : SUCCESS in 46m 11s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-mysql/2df32dc/ : SUCCESS in 30m 54s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-postgresql/ccec5ab/ : SUCCESS in 28m 56s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/9fed401/ : TIMED_OUT in 3h 25m 36s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-undercloud-containers/79875d4/ : SUCCESS in 1h 48m 27s\n- mistral-rally-task http://logs.openstack.org/87/557487/6/check/mistral-rally-task/4f7d9de/ : SUCCESS in 1h 28m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/6/check/openstack-tox-lower-constraints/f46365a/ : SUCCESS in 10m 27s","accounts_in_message":[],"_revision_number":6},{"id":"bc2e0bf8df25da02907236c55203559cb769249f","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2018-10-17 16:33:08.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"bb3001032c438595060da664534c5c301d099164","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-17 22:04:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/6/check/openstack-tox-pep8/d4cae36/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/6/check/openstack-tox-py27/3a06fd0/ : SUCCESS in 9m 32s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/6/check/openstack-tox-py35/c567a64/ : SUCCESS in 10m 05s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/6/check/openstack-tox-py36/148238d/ : SUCCESS in 9m 18s\n- openstack-tox-docs http://logs.openstack.org/87/557487/6/check/openstack-tox-docs/2553d19/html/ : SUCCESS in 5m 25s\n- openstack-tox-cover http://logs.openstack.org/87/557487/6/check/openstack-tox-cover/a07765e/cover/ : SUCCESS in 10m 46s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/6/check/mistral-devstack/6c48577/ : SUCCESS in 47m 43s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/6/check/mistral-devstack-non-apache/596cc35/ : SUCCESS in 48m 37s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/6/check/mistral-devstack-kombu/7d059be/ : SUCCESS in 39m 53s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-mysql/e064668/ : SUCCESS in 31m 23s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-postgresql/609ec6a/ : SUCCESS in 28m 57s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/7d5fbca/ : SUCCESS in 2h 19m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-undercloud-containers/c29e5c2/ : SUCCESS in 1h 52m 28s\n- mistral-rally-task http://logs.openstack.org/87/557487/6/check/mistral-rally-task/ac7e6b0/ : FAILURE in 1h 50m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/6/check/openstack-tox-lower-constraints/2ac6869/ : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":6},{"id":"36bb2b692dbcfc381937802610d21c233edc6b9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-17 22:04:58.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d6af9254cbf9b351e5b427a36b4e9dcd8e8d1af4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-18 00:12:30.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/6/gate/openstack-tox-pep8/ada9e8c/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py27/a402cf5/ : SUCCESS in 8m 44s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py35/34d77e3/ : SUCCESS in 8m 59s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py36/c44008f/ : SUCCESS in 8m 21s\n- openstack-tox-docs http://logs.openstack.org/87/557487/6/gate/openstack-tox-docs/a89fa80/html/ : SUCCESS in 4m 56s\n- mistral-devstack http://logs.openstack.org/87/557487/6/gate/mistral-devstack/6a20b24/ : SUCCESS in 46m 37s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/6/gate/mistral-devstack-non-apache/f85ea5e/ : SUCCESS in 43m 41s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/6/gate/mistral-tox-unit-mysql/0ceb6e5/ : SUCCESS in 30m 10s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/6/gate/mistral-tox-unit-postgresql/70658b6/ : SUCCESS in 28m 39s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/6/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/e87b331/ : FAILURE in 2h 02m 31s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/6/gate/mistral-devstack-kombu/dd2bca7/ : SUCCESS in 50m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/6/gate/openstack-tox-lower-constraints/01d9757/ : FAILURE in 11m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/6/gate/tripleo-ci-centos-7-undercloud-containers/3f5a4ca/ : SUCCESS in 1h 41m 52s","accounts_in_message":[],"_revision_number":6},{"id":"fe9a1e4f18cc5a9cd96e9e4828e6f61224fde9dd","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2018-11-08 16:13:50.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"adf3457d5ce91fc0b482a6a993d820a1744608b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 18:28:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/6/check/openstack-tox-pep8/7b44ec6/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/6/check/openstack-tox-py27/030a4e7/ : SUCCESS in 11m 36s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/6/check/openstack-tox-py35/2a8f43f/ : SUCCESS in 12m 11s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/6/check/openstack-tox-py36/a91401c/ : SUCCESS in 11m 53s\n- openstack-tox-docs http://logs.openstack.org/87/557487/6/check/openstack-tox-docs/6240452/html/ : SUCCESS in 5m 51s\n- openstack-tox-cover http://logs.openstack.org/87/557487/6/check/openstack-tox-cover/7a779a5/cover/ : SUCCESS in 10m 33s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/6/check/mistral-devstack/c857976/ : SUCCESS in 47m 40s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/6/check/mistral-devstack-non-apache/9db4dfa/ : SUCCESS in 47m 14s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/6/check/mistral-devstack-kombu/7a7ea2f/ : SUCCESS in 46m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-mysql/c1d86f1/ : SUCCESS in 31m 56s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/6/check/mistral-tox-unit-postgresql/12baa1d/ : SUCCESS in 30m 08s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/f74c5e5/ : FAILURE in 2h 00m 32s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/6/check/tripleo-ci-centos-7-undercloud-containers/c65f427/ : SUCCESS in 1h 51m 55s\n- mistral-rally-task http://logs.openstack.org/87/557487/6/check/mistral-rally-task/9c9e66d/ : FAILURE in 1h 43m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/6/check/openstack-tox-lower-constraints/e16654a/ : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":6},{"id":"22a8e52d007512b63f7eff06223d09ae59c5ccef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 18:28:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"02b6e8ee2f2900fb98d5b857680f82399ddd4cc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 20:30:13.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/6/gate/openstack-tox-pep8/691e0b6/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py27/a6b8fae/ : SUCCESS in 11m 03s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py35/648483d/ : SUCCESS in 8m 06s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/6/gate/openstack-tox-py36/8ba8c12/ : SUCCESS in 7m 48s\n- openstack-tox-docs http://logs.openstack.org/87/557487/6/gate/openstack-tox-docs/161514b/html/ : SUCCESS in 5m 09s\n- mistral-devstack http://logs.openstack.org/87/557487/6/gate/mistral-devstack/d645a3f/ : SUCCESS in 40m 47s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/6/gate/mistral-devstack-non-apache/80349e6/ : SUCCESS in 48m 14s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/6/gate/mistral-tox-unit-mysql/797aa74/ : SUCCESS in 26m 25s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/6/gate/mistral-tox-unit-postgresql/820dff7/ : SUCCESS in 26m 26s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/6/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/79a4db3/ : FAILURE in 1h 59m 22s (non-voting)\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/6/gate/mistral-devstack-kombu/c23f8c5/ : SUCCESS in 49m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/6/gate/openstack-tox-lower-constraints/f9cd459/ : FAILURE in 10m 54s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/6/gate/tripleo-ci-centos-7-undercloud-containers/029a36f/ : SUCCESS in 1h 44m 40s","accounts_in_message":[],"_revision_number":6},{"id":"0983fc037c0196ef02c5fc6986baa85315117fc3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-13 09:47:41.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"22546c7a86a6254540632bb4d8aaa97189c86b99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 12:33:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/7/check/openstack-tox-pep8/7f484b7/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/7/check/openstack-tox-py27/5498c7c/ : SUCCESS in 7m 51s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/7/check/openstack-tox-py35/89598c3/ : SUCCESS in 8m 07s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/7/check/openstack-tox-py36/6a3ef19/ : SUCCESS in 8m 18s\n- openstack-tox-docs http://logs.openstack.org/87/557487/7/check/openstack-tox-docs/cd08028/html/ : SUCCESS in 4m 20s\n- openstack-tox-cover http://logs.openstack.org/87/557487/7/check/openstack-tox-cover/a16ecab/cover/ : SUCCESS in 8m 38s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/7/check/mistral-devstack/72fd910/ : SUCCESS in 50m 26s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/7/check/mistral-devstack-non-apache/f8b3ad5/ : SUCCESS in 44m 18s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/7/check/mistral-devstack-kombu/37c13fe/ : SUCCESS in 43m 40s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/7/check/mistral-tox-unit-mysql/4f181ec/ : FAILURE in 25m 41s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/7/check/mistral-tox-unit-postgresql/a7142a4/ : SUCCESS in 27m 49s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/7/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/1c0c8db/ : SUCCESS in 2h 39m 32s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/7/check/tripleo-ci-centos-7-undercloud-containers/16f402c/ : SUCCESS in 1h 38m 24s\n- mistral-rally-task http://logs.openstack.org/87/557487/7/check/mistral-rally-task/5666540/ : SUCCESS in 58m 40s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/7/check/openstack-tox-lower-constraints/825bfb5/ : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":7},{"id":"c7bcea98d4cbd1bb9136e9c1a0c164da1903d650","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-19 08:16:43.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"dc1c810f7ce261009b06badb6439822fa2fc0430","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2018-11-19 08:46:48.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"60fba676441ce2279f53021096137f69d496ff1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-19 11:35:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/8/check/openstack-tox-pep8/6bff831/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/8/check/openstack-tox-py27/3d28d37/ : SUCCESS in 8m 34s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/8/check/openstack-tox-py35/806769b/ : SUCCESS in 8m 40s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/8/check/openstack-tox-py36/6be675c/ : SUCCESS in 10m 38s\n- openstack-tox-docs http://logs.openstack.org/87/557487/8/check/openstack-tox-docs/6a8dd40/html/ : SUCCESS in 5m 39s\n- openstack-tox-cover http://logs.openstack.org/87/557487/8/check/openstack-tox-cover/bb8cb41/cover/ : SUCCESS in 9m 25s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/8/check/mistral-devstack/6eb81c1/ : SUCCESS in 41m 26s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/8/check/mistral-devstack-non-apache/a5d5ff8/ : SUCCESS in 48m 19s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/8/check/mistral-devstack-kombu/f93ab6e/ : SUCCESS in 37m 07s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/8/check/mistral-tox-unit-mysql/099b1ba/ : SUCCESS in 27m 03s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/8/check/mistral-tox-unit-postgresql/557f065/ : SUCCESS in 25m 34s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/8/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/f701d80/ : TIMED_OUT in 3h 16m 50s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/8/check/tripleo-ci-centos-7-undercloud-containers/0b2402f/ : SUCCESS in 2h 11m 39s\n- mistral-rally-task http://logs.openstack.org/87/557487/8/check/mistral-rally-task/bb90ae1/ : SUCCESS in 59m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/8/check/openstack-tox-lower-constraints/f2042d3/ : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":8},{"id":"dd6098059dc044d61f1925e991c5f5056ec36a19","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-27 05:55:50.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"378a19cc56cac402039865929bee32c7e9306cd3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2018-11-27 05:55:55.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"641e8b444b0ec137e7c0196b3193d122b4fd86cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 08:13:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/9/check/openstack-tox-pep8/33efc80/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/9/check/openstack-tox-py27/6853362/ : SUCCESS in 9m 45s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/9/check/openstack-tox-py35/e3b8b2d/ : SUCCESS in 9m 19s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/9/check/openstack-tox-py36/3d33ad6/ : SUCCESS in 10m 02s\n- openstack-tox-docs http://logs.openstack.org/87/557487/9/check/openstack-tox-docs/1a3c760/html/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/87/557487/9/check/openstack-tox-cover/b001e91/cover/ : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/9/check/mistral-devstack/c53551f/ : SUCCESS in 39m 08s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/9/check/mistral-devstack-non-apache/a760b73/ : SUCCESS in 54m 59s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/9/check/mistral-devstack-kombu/49f27da/ : SUCCESS in 43m 48s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/9/check/mistral-tox-unit-mysql/fd5cfd3/ : FAILURE in 35m 02s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/9/check/mistral-tox-unit-postgresql/4b56612/ : SUCCESS in 27m 46s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/9/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/8f864da/ : SUCCESS in 2h 15m 59s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/9/check/tripleo-ci-centos-7-undercloud-containers/7e347da/ : SUCCESS in 1h 32m 48s\n- mistral-rally-task http://logs.openstack.org/87/557487/9/check/mistral-rally-task/500ec0c/ : SUCCESS in 55m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/9/check/openstack-tox-lower-constraints/b1fc227/ : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":9},{"id":"dfd3c7425e95967e0dfe81bcc8c726db34f105b3","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2018-11-27 08:24:50.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"32b151b6ce23bfff2c15726a727b5acdc0eabd5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 10:48:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/9/check/openstack-tox-pep8/b1d6685/ : SUCCESS in 6m 11s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/9/check/openstack-tox-py27/a7485e5/ : SUCCESS in 8m 40s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/9/check/openstack-tox-py35/f172b42/ : SUCCESS in 8m 41s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/9/check/openstack-tox-py36/2a65431/ : SUCCESS in 8m 23s\n- openstack-tox-docs http://logs.openstack.org/87/557487/9/check/openstack-tox-docs/66688e9/html/ : SUCCESS in 4m 34s\n- openstack-tox-cover http://logs.openstack.org/87/557487/9/check/openstack-tox-cover/8d4652e/cover/ : SUCCESS in 10m 13s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/9/check/mistral-devstack/47924f4/ : SUCCESS in 41m 54s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/9/check/mistral-devstack-non-apache/d56d235/ : SUCCESS in 43m 44s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/9/check/mistral-devstack-kombu/0e3cfb6/ : SUCCESS in 1h 15m 24s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/9/check/mistral-tox-unit-mysql/5596163/ : SUCCESS in 28m 31s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/9/check/mistral-tox-unit-postgresql/d9085e6/ : SUCCESS in 25m 34s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/9/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/9df7fee/ : SUCCESS in 2h 21m 33s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/9/check/tripleo-ci-centos-7-undercloud-containers/b054937/ : SUCCESS in 1h 39m 48s\n- mistral-rally-task http://logs.openstack.org/87/557487/9/check/mistral-rally-task/6261d66/ : SUCCESS in 1h 09m 35s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/9/check/openstack-tox-lower-constraints/be78534/ : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":9},{"id":"2632b3d15765fa19d1132230fede43183ff56c9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 10:49:11.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4e6da0a762572bce2263f102d17a8126d3c43b42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 13:30:45.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/9/gate/openstack-tox-pep8/356a999/ : SUCCESS in 14m 14s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/9/gate/openstack-tox-py27/169bff2/ : SUCCESS in 12m 50s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/9/gate/openstack-tox-py35/eb8be85/ : SUCCESS in 11m 15s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/9/gate/openstack-tox-py36/60f89ff/ : SUCCESS in 9m 04s\n- openstack-tox-docs http://logs.openstack.org/87/557487/9/gate/openstack-tox-docs/7ab382f/html/ : SUCCESS in 6m 50s\n- mistral-devstack http://logs.openstack.org/87/557487/9/gate/mistral-devstack/9ba4384/ : SUCCESS in 44m 49s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/9/gate/mistral-devstack-non-apache/2158562/ : SUCCESS in 47m 11s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/9/gate/mistral-tox-unit-mysql/27f4f37/ : FAILURE in 29m 35s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/9/gate/mistral-tox-unit-postgresql/9ad8667/ : SUCCESS in 26m 31s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/87/557487/9/gate/tripleo-ci-centos-7-scenario003-multinode-oooq-container/6227dbf/ : SUCCESS in 2h 34m 19s (non-voting)\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/9/gate/mistral-devstack-kombu/0be4d18/ : SUCCESS in 1h 10m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/9/gate/openstack-tox-lower-constraints/3a59067/ : FAILURE in 11m 46s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/9/gate/tripleo-ci-centos-7-undercloud-containers/4b4da25/ : SUCCESS in 1h 39m 23s","accounts_in_message":[],"_revision_number":9},{"id":"a628bce59215baa1a0f6c95948f4fa7fe306b431","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-01-11 05:07:08.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"33caf0eb1e699dec32700df93976b1a1369978af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 06:37:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/10/check/openstack-tox-pep8/ca44888/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/10/check/openstack-tox-py27/0fd4640/ : FAILURE in 11m 00s\n- openstack-tox-py35 http://logs.openstack.org/87/557487/10/check/openstack-tox-py35/b9d95ea/ : FAILURE in 9m 51s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/10/check/openstack-tox-py36/bb528c9/ : FAILURE in 7m 50s\n- openstack-tox-docs http://logs.openstack.org/87/557487/10/check/openstack-tox-docs/3fccc1e/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/87/557487/10/check/openstack-tox-cover/6069d7c/ : FAILURE in 11m 34s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/10/check/mistral-devstack/80cb238/ : SUCCESS in 35m 40s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/10/check/mistral-devstack-non-apache/2c6e31d/ : SUCCESS in 33m 18s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/10/check/mistral-devstack-kombu/9d3dfa8/ : SUCCESS in 34m 59s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/10/check/mistral-tox-unit-mysql/78dad1d/ : FAILURE in 30m 11s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/10/check/mistral-tox-unit-postgresql/8c016a9/ : FAILURE in 27m 23s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/10/check/tripleo-ci-centos-7-undercloud-containers/8959f32/ : SUCCESS in 1h 29m 35s\n- mistral-rally-task http://logs.openstack.org/87/557487/10/check/mistral-rally-task/480ddfb/ : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/10/check/openstack-tox-lower-constraints/972c002/ : FAILURE in 11m 29s","accounts_in_message":[],"_revision_number":10},{"id":"3f81322ac954888827f58cbc1534fb125bc45833","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-09 07:36:16.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"bb1ec36bc1b3b4ca6f47887d151c86d18d8b1eb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 10:14:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/11/check/openstack-tox-pep8/5af76d6/ : SUCCESS in 8m 15s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/11/check/openstack-tox-py27/9e5ce94/ : SUCCESS in 11m 49s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/11/check/openstack-tox-py36/d0dfbbd/ : SUCCESS in 9m 55s\n- openstack-tox-py37 http://logs.openstack.org/87/557487/11/check/openstack-tox-py37/813c4db/ : SUCCESS in 8m 46s\n- openstack-tox-docs http://logs.openstack.org/87/557487/11/check/openstack-tox-docs/c662d3d/html/ : SUCCESS in 5m 40s\n- openstack-tox-cover http://logs.openstack.org/87/557487/11/check/openstack-tox-cover/4e0f9c9/cover/ : SUCCESS in 9m 30s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/11/check/mistral-devstack/112a0c5/ : SUCCESS in 39m 22s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/11/check/mistral-devstack-non-apache/8d3e576/ : SUCCESS in 36m 45s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/11/check/mistral-devstack-kombu/3ec2330/ : SUCCESS in 40m 46s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/11/check/mistral-tox-unit-mysql/e31e9a1/ : SUCCESS in 28m 35s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/11/check/mistral-tox-unit-postgresql/9c9cf0b/ : SUCCESS in 29m 51s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/87/557487/11/check/tripleo-ci-centos-7-scenario003-standalone/38d206b/ : SUCCESS in 1h 15m 33s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/87/557487/11/check/tripleo-ci-centos-7-containers-multinode/cda2b92/ : SUCCESS in 1h 56m 55s\n- mistral-rally-task http://logs.openstack.org/87/557487/11/check/mistral-rally-task/3a350ae/ : FAILURE in 1h 02m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/11/check/openstack-tox-lower-constraints/2be9d94/ : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":11},{"id":"23bcb08187d16dadc294ebc734ae066881ca1088","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-09 10:15:29.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"e19cd78541577a87cafef4d23d0fa1c6d8aefa0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 10:15:41.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"98bbcbe11eb2df10c65708f5dc0a435c01c77936","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-06-09 10:17:19.000000000","message":"Patch Set 11:\n\nThe sad news is that the Kombi driver is not production quality and should not be used in deployments. Does this patch pertain to oslo_messaging too? This is the only way currently to make use of this great content in production.","accounts_in_message":[],"_revision_number":11},{"id":"960e63bc42923529cc4f2bcd558b3ec3437416f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 11:40:26.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/11/gate/openstack-tox-pep8/3a3d67d/ : SUCCESS in 5m 17s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/11/gate/openstack-tox-py27/982817e/ : SUCCESS in 10m 43s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/11/gate/openstack-tox-py36/ff4d0be/ : SUCCESS in 12m 11s\n- openstack-tox-py37 http://logs.openstack.org/87/557487/11/gate/openstack-tox-py37/29f6467/ : SUCCESS in 11m 31s\n- openstack-tox-docs http://logs.openstack.org/87/557487/11/gate/openstack-tox-docs/257564f/html/ : SUCCESS in 6m 56s\n- mistral-devstack http://logs.openstack.org/87/557487/11/gate/mistral-devstack/c0d902e/ : SUCCESS in 46m 08s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/11/gate/mistral-devstack-non-apache/56597c0/ : SUCCESS in 43m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/11/gate/mistral-tox-unit-mysql/d6dc0da/ : SUCCESS in 26m 52s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/11/gate/mistral-tox-unit-postgresql/7d7cef8/ : SUCCESS in 28m 57s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/87/557487/11/gate/tripleo-ci-centos-7-scenario003-standalone/393f785/ : SUCCESS in 1h 17m 08s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/11/gate/mistral-devstack-kombu/6250e69/ : SUCCESS in 35m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/11/gate/openstack-tox-lower-constraints/d1ec564/ : FAILURE in 12m 31s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/11/gate/tripleo-ci-centos-7-undercloud-containers/fddd50b/ : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":11},{"id":"c4c8d9cf0de8d03b1e8b40713fe5b81c6d00ffb7","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-09 15:38:52.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"a8f19edbe917a1407ebf33479c02616fce8b078b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 17:51:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/11/check/openstack-tox-pep8/0bd071d/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/11/check/openstack-tox-py27/3411e21/ : SUCCESS in 10m 06s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/11/check/openstack-tox-py36/77593e5/ : SUCCESS in 8m 51s\n- openstack-tox-py37 http://logs.openstack.org/87/557487/11/check/openstack-tox-py37/901dc6f/ : SUCCESS in 8m 47s\n- openstack-tox-docs http://logs.openstack.org/87/557487/11/check/openstack-tox-docs/109336d/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/87/557487/11/check/openstack-tox-cover/2fbf136/cover/ : SUCCESS in 8m 55s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/11/check/mistral-devstack/2548f5d/ : SUCCESS in 45m 05s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/11/check/mistral-devstack-non-apache/f994bff/ : SUCCESS in 43m 44s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/11/check/mistral-devstack-kombu/d1dec87/ : SUCCESS in 44m 08s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/11/check/mistral-tox-unit-mysql/199bd9a/ : SUCCESS in 28m 38s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/11/check/mistral-tox-unit-postgresql/8c41159/ : SUCCESS in 26m 00s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/87/557487/11/check/tripleo-ci-centos-7-scenario003-standalone/12436c1/ : SUCCESS in 1h 10m 29s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/87/557487/11/check/tripleo-ci-centos-7-containers-multinode/7871ba5/ : SUCCESS in 1h 58m 55s\n- mistral-rally-task http://logs.openstack.org/87/557487/11/check/mistral-rally-task/c54e9c3/ : SUCCESS in 1h 04m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/11/check/openstack-tox-lower-constraints/ab62c16/ : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":11},{"id":"a4a737a8a4992e1076d6681fc9a48a6a4a5d5b00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 17:51:24.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"e13172e7bbad131885149eba75b6127f1e97b74a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 19:38:14.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/11/gate/openstack-tox-pep8/a25783a/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/11/gate/openstack-tox-py27/922de01/ : SUCCESS in 8m 04s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/11/gate/openstack-tox-py36/0a3c0e8/ : SUCCESS in 8m 22s\n- openstack-tox-py37 http://logs.openstack.org/87/557487/11/gate/openstack-tox-py37/d226e30/ : SUCCESS in 8m 38s\n- openstack-tox-docs http://logs.openstack.org/87/557487/11/gate/openstack-tox-docs/b4ff17e/html/ : SUCCESS in 4m 34s\n- mistral-devstack http://logs.openstack.org/87/557487/11/gate/mistral-devstack/7e65bf3/ : SUCCESS in 41m 51s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/11/gate/mistral-devstack-non-apache/407d579/ : SUCCESS in 42m 53s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/11/gate/mistral-tox-unit-mysql/f8a4e18/ : SUCCESS in 26m 55s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/11/gate/mistral-tox-unit-postgresql/ef602ab/ : SUCCESS in 26m 02s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/87/557487/11/gate/tripleo-ci-centos-7-scenario003-standalone/be2bddc/ : SUCCESS in 1h 11m 57s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/11/gate/mistral-devstack-kombu/5869ea6/ : SUCCESS in 40m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/11/gate/openstack-tox-lower-constraints/ff93edd/ : FAILURE in 9m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/11/gate/tripleo-ci-centos-7-undercloud-containers/724bfed/ : SUCCESS in 1h 41m 04s","accounts_in_message":[],"_revision_number":11},{"id":"327afd2f83130bc1be6e00cee2e8f4126c4112f1","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-14 21:25:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6e81034ef42ecff9df2ebfc5af8102bd736e87c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:05:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/12/check/openstack-tox-pep8/e356de8/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/12/check/openstack-tox-py27/c58f9e3/ : SUCCESS in 9m 02s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/12/check/openstack-tox-py36/32cc273/ : SUCCESS in 9m 52s\n- openstack-tox-py37 http://logs.openstack.org/87/557487/12/check/openstack-tox-py37/217ce79/ : SUCCESS in 9m 07s\n- openstack-tox-docs http://logs.openstack.org/87/557487/12/check/openstack-tox-docs/daa2bb5/html/ : SUCCESS in 5m 14s\n- requirements-check http://logs.openstack.org/87/557487/12/check/requirements-check/5c3cf6f/ : FAILURE in 3m 34s\n- openstack-tox-cover http://logs.openstack.org/87/557487/12/check/openstack-tox-cover/ea300ce/cover/ : SUCCESS in 9m 25s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/12/check/mistral-devstack/d813964/ : SUCCESS in 44m 23s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/12/check/mistral-devstack-non-apache/c9401aa/ : SUCCESS in 44m 19s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/12/check/mistral-devstack-kombu/4f4e7be/ : SUCCESS in 46m 20s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/12/check/mistral-tox-unit-mysql/528b03f/ : SUCCESS in 28m 03s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/12/check/mistral-tox-unit-postgresql/90cbc6d/ : SUCCESS in 26m 48s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/87/557487/12/check/tripleo-ci-centos-7-scenario003-standalone/bb3d5f1/ : SUCCESS in 1h 14m 30s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/87/557487/12/check/tripleo-ci-centos-7-containers-multinode/230d11e/ : SUCCESS in 2h 06m 37s\n- mistral-rally-task http://logs.openstack.org/87/557487/12/check/mistral-rally-task/dec417d/ : SUCCESS in 1h 07m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/12/check/openstack-tox-lower-constraints/2f26463/ : SUCCESS in 10m 58s","accounts_in_message":[],"_revision_number":12},{"id":"51825fae64948fbcc7147c7e4c8a4da443443a94","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-06-17 06:55:41.000000000","message":"Patch Set 12: Code-Review+1\n\nI think it requirements.txt needs to be changed on line 26\nbump up \noslo.utils\u003e\u003d3.33.0 # Apache-2.0","accounts_in_message":[],"_revision_number":12},{"id":"6ba1647ee2e27629173f3e003bb2909948146976","author":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"date":"2019-06-17 11:28:25.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"66bdfbfdf649facd146b8233ba73140389690582","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:03:55.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/13/check/openstack-tox-pep8/b86be7c/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/13/check/openstack-tox-py27/9f5045c/ : SUCCESS in 8m 43s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/13/check/openstack-tox-py36/d8a93ad/ : SUCCESS in 9m 00s\n- openstack-tox-py37 http://logs.openstack.org/87/557487/13/check/openstack-tox-py37/d99d2af/ : SUCCESS in 9m 06s\n- openstack-tox-docs http://logs.openstack.org/87/557487/13/check/openstack-tox-docs/2b6e488/html/ : SUCCESS in 5m 03s\n- requirements-check http://logs.openstack.org/87/557487/13/check/requirements-check/ccf512e/ : SUCCESS in 3m 29s\n- openstack-tox-cover http://logs.openstack.org/87/557487/13/check/openstack-tox-cover/ed025e6/cover/ : SUCCESS in 9m 22s (non-voting)\n- mistral-devstack http://logs.openstack.org/87/557487/13/check/mistral-devstack/489a1ad/ : SUCCESS in 43m 40s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/13/check/mistral-devstack-non-apache/4e2fc90/ : SUCCESS in 43m 32s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/13/check/mistral-devstack-kombu/6add9bd/ : SUCCESS in 44m 51s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/13/check/mistral-tox-unit-mysql/1bf2381/ : SUCCESS in 27m 36s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/13/check/mistral-tox-unit-postgresql/4d9e00b/ : SUCCESS in 27m 25s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/87/557487/13/check/tripleo-ci-centos-7-scenario003-standalone/5918aca/ : SUCCESS in 1h 15m 05s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/87/557487/13/check/tripleo-ci-centos-7-containers-multinode/dc7857c/ : SUCCESS in 2h 10m 48s\n- mistral-rally-task http://logs.openstack.org/87/557487/13/check/mistral-rally-task/4ce06cb/ : SUCCESS in 1h 17m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/13/check/openstack-tox-lower-constraints/a474214/ : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":13},{"id":"bcbc87c7a1be7c7b3986f690ef22616440d0dfee","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-06-17 14:59:29.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"3043b673ea180aafe267d64a4fa95058ed014f2b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-06-18 04:54:57.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"d6a7aeed1ff466156c8702335620ad7f89088bb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 04:55:09.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"7b81e158b467ad2525d1587628e72e2c982bd57b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 06:35:19.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/557487/13/gate/openstack-tox-pep8/fbecb64/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/87/557487/13/gate/openstack-tox-py27/e2fb1ae/ : SUCCESS in 8m 53s\n- openstack-tox-py36 http://logs.openstack.org/87/557487/13/gate/openstack-tox-py36/a882989/ : SUCCESS in 10m 00s\n- openstack-tox-py37 http://logs.openstack.org/87/557487/13/gate/openstack-tox-py37/fd02e75/ : SUCCESS in 9m 49s\n- openstack-tox-docs http://logs.openstack.org/87/557487/13/gate/openstack-tox-docs/49a50e4/html/ : SUCCESS in 4m 48s\n- requirements-check http://logs.openstack.org/87/557487/13/gate/requirements-check/7cffb44/ : SUCCESS in 4m 57s\n- mistral-devstack http://logs.openstack.org/87/557487/13/gate/mistral-devstack/0ea317d/ : SUCCESS in 36m 58s\n- mistral-devstack-non-apache http://logs.openstack.org/87/557487/13/gate/mistral-devstack-non-apache/f63da44/ : SUCCESS in 40m 45s\n- mistral-tox-unit-mysql http://logs.openstack.org/87/557487/13/gate/mistral-tox-unit-mysql/892649d/ : SUCCESS in 28m 57s\n- mistral-tox-unit-postgresql http://logs.openstack.org/87/557487/13/gate/mistral-tox-unit-postgresql/5e78a5a/ : SUCCESS in 26m 34s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/87/557487/13/gate/tripleo-ci-centos-7-scenario003-standalone/ec0620b/ : SUCCESS in 1h 15m 01s\n- mistral-devstack-kombu http://logs.openstack.org/87/557487/13/gate/mistral-devstack-kombu/29203a6/ : SUCCESS in 38m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/87/557487/13/gate/openstack-tox-lower-constraints/8417298/ : SUCCESS in 11m 48s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/87/557487/13/gate/tripleo-ci-centos-7-undercloud-containers/ca770be/ : SUCCESS in 1h 32m 49s","accounts_in_message":[],"_revision_number":13},{"id":"178d7244583164b378c323c8c92226624acb2039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 06:35:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"b97b24856f611998ade3ac7390074dc02b880b19","revisions":{"799452a51fe958555952e0e5d38d5f9dfc2e58ac":{"kind":"REWORK","_number":1,"created":"2018-03-28 19:07:15.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/87/557487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/1"}}},"commit":{"parents":[{"commit":"777ba58710803bf291506c1539596b60660f7fbc","subject":"Merge \"Add a release note for the \u0027pause\u0027 command fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/777ba58710803bf291506c1539596b60660f7fbc"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/799452a51fe958555952e0e5d38d5f9dfc2e58ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/799452a51fe958555952e0e5d38d5f9dfc2e58ac"}]},"branch":"refs/heads/master"},"5a45e207d9b827bb594457234312072f59924f09":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2018-04-04 18:58:20.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/87/557487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/2"}}},"commit":{"parents":[{"commit":"54fe5b0109a8a87614e1c0fa505b13c491c3f759","subject":"Merge \"Fix WF execution getting stuck on null description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54fe5b0109a8a87614e1c0fa505b13c491c3f759"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-04-04 18:57:52.000000000","tz":-240},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5a45e207d9b827bb594457234312072f59924f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5a45e207d9b827bb594457234312072f59924f09"}]},"branch":"refs/heads/master"},"05e1e21e43832998bd0f5c2c34e5d8efd801165d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-09-25 14:15:07.000000000","uploader":{"_account_id":7065,"name":"Brad P. Crochet","email":"brad@redhat.com","username":"bcrochet"},"ref":"refs/changes/87/557487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/3"}}},"commit":{"parents":[{"commit":"8720a2711b736b87ff4bf96239a52e278b0b6fa2","subject":"Merge \"Fix how Mistral calculates workflow output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8720a2711b736b87ff4bf96239a52e278b0b6fa2"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Brad P. Crochet","email":"brad@redhat.com","date":"2018-09-25 14:15:07.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/05e1e21e43832998bd0f5c2c34e5d8efd801165d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/05e1e21e43832998bd0f5c2c34e5d8efd801165d"}]},"branch":"refs/heads/master"},"de8573c615e01caebcc7b2132601ea4dc5fda69e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-10-02 11:13:17.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/87/557487/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/4"}}},"commit":{"parents":[{"commit":"f0b49a82b4c4d00961f051f69470325c8c481999","subject":"Add a release note for Ic98e2db02abd8483591756d73e06784cc2e9cbe3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0b49a82b4c4d00961f051f69470325c8c481999"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2018-10-02 11:13:17.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de8573c615e01caebcc7b2132601ea4dc5fda69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de8573c615e01caebcc7b2132601ea4dc5fda69e"}]},"branch":"refs/heads/master"},"cfa2a0f66823269600ba5b40c15176d9783b2bb9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-10-08 12:02:39.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/87/557487/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/5"}}},"commit":{"parents":[{"commit":"fc9876ec7bb1a3dccbc76656bf7a92fd59b0a38b","subject":"Merge \"Make task execution logging more readable and informative\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fc9876ec7bb1a3dccbc76656bf7a92fd59b0a38b"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2018-10-08 12:02:39.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cfa2a0f66823269600ba5b40c15176d9783b2bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cfa2a0f66823269600ba5b40c15176d9783b2bb9"}]},"branch":"refs/heads/master"},"fdee1095a7c0d73f0afd3953e88ba558e3daf18d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-10-14 10:28:24.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/87/557487/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/6"}}},"commit":{"parents":[{"commit":"9be7e928d683ff00aaa6617c5ba57456fe6a02cc","subject":"Remove remaining references to the rpc_backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9be7e928d683ff00aaa6617c5ba57456fe6a02cc"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-10-14 10:28:24.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdee1095a7c0d73f0afd3953e88ba558e3daf18d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fdee1095a7c0d73f0afd3953e88ba558e3daf18d"}]},"branch":"refs/heads/master"},"9d53035572a6d1d15f552c83697959d45c2dbb20":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2018-11-13 09:47:41.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/87/557487/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/7"}}},"commit":{"parents":[{"commit":"4b83cef3af8586463898ed165db4866564dec48c","subject":"Merge \"Divide yaml input to save it into definitions separately.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4b83cef3af8586463898ed165db4866564dec48c"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-11-13 09:47:41.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9d53035572a6d1d15f552c83697959d45c2dbb20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9d53035572a6d1d15f552c83697959d45c2dbb20"}]},"branch":"refs/heads/master"},"076c96df9cd76b623c489ff21b35a29bcbb1d32d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2018-11-19 08:16:43.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/87/557487/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/8"}}},"commit":{"parents":[{"commit":"c9e08a883946540bf245dd372406642e3aa50808","subject":"Fix \"join\" when the last indirect inbound task failed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c9e08a883946540bf245dd372406642e3aa50808"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-11-19 08:16:43.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/076c96df9cd76b623c489ff21b35a29bcbb1d32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/076c96df9cd76b623c489ff21b35a29bcbb1d32d"}]},"branch":"refs/heads/master"},"2ad0a2ab6709710851407f53ab868271de10ea1d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-11-27 05:55:50.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/87/557487/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/9"}}},"commit":{"parents":[{"commit":"3153f1ca584450bdd1b983b8ebc13ef5cb809c0e","subject":"Set admin security context in the action execution checker thread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3153f1ca584450bdd1b983b8ebc13ef5cb809c0e"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2018-11-27 05:55:50.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ad0a2ab6709710851407f53ab868271de10ea1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ad0a2ab6709710851407f53ab868271de10ea1d"}]},"branch":"refs/heads/master"},"aa2ea0bddee64f2993a2e0ea3bac8f3221630ced":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-01-11 05:07:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/87/557487/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/10"}}},"commit":{"parents":[{"commit":"c7509a9c858ffeef9a3868b708b9335c6258618a","subject":"Fix the misspelling of \"default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c7509a9c858ffeef9a3868b708b9335c6258618a"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-01-11 05:07:08.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aa2ea0bddee64f2993a2e0ea3bac8f3221630ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aa2ea0bddee64f2993a2e0ea3bac8f3221630ced"}]},"branch":"refs/heads/master"},"1b0c082d1bbed28fe212344e422dee22490ba944":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-06-09 07:36:16.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/87/557487/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/11"}}},"commit":{"parents":[{"commit":"5131444a62a37fb9a6b9640d5bddd49b6578c89e","subject":"Merge \"Use get_task_executions_count for any_cancels method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5131444a62a37fb9a6b9640d5bddd49b6578c89e"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-09 07:36:16.000000000","tz":0},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1b0c082d1bbed28fe212344e422dee22490ba944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1b0c082d1bbed28fe212344e422dee22490ba944"}]},"branch":"refs/heads/master"},"44965cc3f7d7465823ea9ab5486cfbdc3a33eb84":{"kind":"REWORK","_number":12,"created":"2019-06-14 21:25:39.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/87/557487/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/12"}}},"commit":{"parents":[{"commit":"5131444a62a37fb9a6b9640d5bddd49b6578c89e","subject":"Merge \"Use get_task_executions_count for any_cancels method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5131444a62a37fb9a6b9640d5bddd49b6578c89e"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-14 21:25:03.000000000","tz":120},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/44965cc3f7d7465823ea9ab5486cfbdc3a33eb84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/44965cc3f7d7465823ea9ab5486cfbdc3a33eb84"}]},"branch":"refs/heads/master"},"b97b24856f611998ade3ac7390074dc02b880b19":{"kind":"REWORK","_number":13,"created":"2019-06-17 11:28:25.000000000","uploader":{"_account_id":11391,"name":"Mikhail Fedosin","email":"mfedosin@redhat.com","username":"fedosinme"},"ref":"refs/changes/87/557487/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/87/557487/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/87/557487/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/87/557487/13"}}},"commit":{"parents":[{"commit":"5131444a62a37fb9a6b9640d5bddd49b6578c89e","subject":"Merge \"Use get_task_executions_count for any_cancels method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5131444a62a37fb9a6b9640d5bddd49b6578c89e"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2018-03-28 19:01:36.000000000","tz":-240},"committer":{"name":"Mike Fedosin","email":"mfedosin@redhat.com","date":"2019-06-17 11:27:55.000000000","tz":120},"subject":"Use eventlet-aware threading events","message":"Use eventlet-aware threading events\n\nFor the gory details behind this, see:\n\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1384183\n\nIn summary, if you use threading.Event on python2 with eventlet, it\nunnecessarily spams system calls to epoll_wait and burns a *lot* of\nCPU cycles by constantly context-switching.\n\nI added eventletutils.Event to oslo_utils a while back to\nintelligently use the eventlet Event object when eventlet is being\nused, and to fall back to the threading one otherwise.  So just use\nthat and it will do the right thing.\n\nDepends-On: https://review.openstack.org/#/c/558879/\nChange-Id: I6ca0d0fc2a2ff393bab502a5f705fb0d63c32c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b97b24856f611998ade3ac7390074dc02b880b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b97b24856f611998ade3ac7390074dc02b880b19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
