)]}'
{"id":"openstack%2Fmistral~565906","triplet_id":"openstack%2Fmistral~master~I5ae6a922e1fde60ab80ebd7ca81a5aff5ca2df9b","project":"openstack/mistral","branch":"master","topic":"bug/1767352","hashtags":[],"change_id":"I5ae6a922e1fde60ab80ebd7ca81a5aff5ca2df9b","subject":"Timeout policy must fail task without retry","status":"ABANDONED","created":"2018-05-02 22:22:43.000000000","updated":"2026-03-28 19:37:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b97bce1b704419c957e8cda3cdd3155e2571ec75","_number":565906,"virtual_id_number":565906,"owner":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":-1,"date":"2019-04-08 11:38:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-03 07:01:30.000000000","updated_by":{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},"reviewer":{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},"state":"REVIEWER"},{"updated":"2019-03-27 20:23:46.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-04-08 11:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"624e4c951524e2fb0c183752e1a684962ff77c4c","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-05-02 22:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23eff99e4dfd7dd3cd1ff5cc424d0bfa8013d3f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 00:35:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/06/565906/1/check/openstack-tox-pep8/f1ad796/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/06/565906/1/check/openstack-tox-py27/c9f0b28/ : SUCCESS in 8m 37s\n- openstack-tox-py35 http://logs.openstack.org/06/565906/1/check/openstack-tox-py35/06ec2b8/ : SUCCESS in 9m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/565906/1/check/build-openstack-sphinx-docs/c679c23/html/ : SUCCESS in 4m 59s\n- openstack-tox-cover http://logs.openstack.org/06/565906/1/check/openstack-tox-cover/aeb4877/ : FAILURE in 12m 29s (non-voting)\n- mistral-devstack http://logs.openstack.org/06/565906/1/check/mistral-devstack/ab1852e/ : SUCCESS in 41m 18s\n- mistral-devstack-non-apache http://logs.openstack.org/06/565906/1/check/mistral-devstack-non-apache/dd0a99e/ : SUCCESS in 39m 14s\n- mistral-devstack-kombu http://logs.openstack.org/06/565906/1/check/mistral-devstack-kombu/9be846c/ : SUCCESS in 39m 15s\n- mistral-tox-unit-mysql http://logs.openstack.org/06/565906/1/check/mistral-tox-unit-mysql/b0b1caf/ : FAILURE in 27m 11s\n- mistral-tox-unit-postgresql http://logs.openstack.org/06/565906/1/check/mistral-tox-unit-postgresql/12f7a4e/ : SUCCESS in 27m 34s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/06/565906/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/885aef5/ : SUCCESS in 2h 10m 43s\n- mistral-rally-task http://logs.openstack.org/06/565906/1/check/mistral-rally-task/7f8754f/ : SUCCESS in 1h 16m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/565906/1/check/openstack-tox-lower-constraints/85cab2d/ : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6b5da1e2eae18d44f9817fd06ca6b7ba072c5b7b","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-05-03 02:44:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a110eab8850a5c8fe6fc135a665a829f6e03105","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 05:08:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/06/565906/2/check/openstack-tox-pep8/b4bc3c6/ : SUCCESS in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/06/565906/2/check/openstack-tox-py27/1104f9d/ : SUCCESS in 9m 19s\n- openstack-tox-py35 http://logs.openstack.org/06/565906/2/check/openstack-tox-py35/593b251/ : SUCCESS in 10m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/565906/2/check/build-openstack-sphinx-docs/62c9607/html/ : SUCCESS in 5m 15s\n- openstack-tox-cover http://logs.openstack.org/06/565906/2/check/openstack-tox-cover/a281d76/ : FAILURE in 12m 07s (non-voting)\n- mistral-devstack http://logs.openstack.org/06/565906/2/check/mistral-devstack/9c65c0c/ : SUCCESS in 39m 50s\n- mistral-devstack-non-apache http://logs.openstack.org/06/565906/2/check/mistral-devstack-non-apache/f0e3473/ : SUCCESS in 39m 37s\n- mistral-devstack-kombu http://logs.openstack.org/06/565906/2/check/mistral-devstack-kombu/41a0e35/ : SUCCESS in 39m 44s\n- mistral-tox-unit-mysql http://logs.openstack.org/06/565906/2/check/mistral-tox-unit-mysql/0b778bb/ : SUCCESS in 27m 47s\n- mistral-tox-unit-postgresql http://logs.openstack.org/06/565906/2/check/mistral-tox-unit-postgresql/fa1bd6e/ : SUCCESS in 27m 02s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/06/565906/2/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/deea240/ : SUCCESS in 2h 17m 53s\n- mistral-rally-task http://logs.openstack.org/06/565906/2/check/mistral-rally-task/028d103/ : SUCCESS in 57m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/565906/2/check/openstack-tox-lower-constraints/a04b08e/ : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":2},{"id":"bcec61af2852f3e6289df91c9480edf929a587fd","author":{"_account_id":25167,"name":"Qi Peng","email":"19921207pq@gmail.com","username":"pengdake"},"date":"2018-05-03 07:01:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5e82733656fdaea5c988e0177ce0311bce220687","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-05-06 10:31:29.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2fb46d6ec8d259225206f2c01f6e8b97ee6ab857","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-06 12:51:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/06/565906/3/check/openstack-tox-pep8/4fc4127/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/06/565906/3/check/openstack-tox-py27/df44d2d/ : SUCCESS in 8m 02s\n- openstack-tox-py35 http://logs.openstack.org/06/565906/3/check/openstack-tox-py35/3777571/ : SUCCESS in 8m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/565906/3/check/build-openstack-sphinx-docs/64a8f73/html/ : SUCCESS in 4m 14s\n- openstack-tox-cover http://logs.openstack.org/06/565906/3/check/openstack-tox-cover/89c9b27/cover/ : SUCCESS in 12m 14s (non-voting)\n- mistral-devstack http://logs.openstack.org/06/565906/3/check/mistral-devstack/e6d3767/ : SUCCESS in 40m 43s\n- mistral-devstack-non-apache http://logs.openstack.org/06/565906/3/check/mistral-devstack-non-apache/79b9096/ : SUCCESS in 41m 32s\n- mistral-devstack-kombu http://logs.openstack.org/06/565906/3/check/mistral-devstack-kombu/c8091f9/ : SUCCESS in 38m 14s\n- mistral-tox-unit-mysql http://logs.openstack.org/06/565906/3/check/mistral-tox-unit-mysql/34f7aca/ : FAILURE in 26m 25s\n- mistral-tox-unit-postgresql http://logs.openstack.org/06/565906/3/check/mistral-tox-unit-postgresql/83e3e49/ : SUCCESS in 25m 46s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/06/565906/3/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/6bbee84/ : SUCCESS in 2h 14m 24s\n- mistral-rally-task http://logs.openstack.org/06/565906/3/check/mistral-rally-task/6c3ee8f/ : SUCCESS in 58m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/565906/3/check/openstack-tox-lower-constraints/a3d7a4b/ : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":3},{"id":"1ee771545d5d067f8f069101faceb1067b6da687","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-03-27 20:23:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f545b3482f35ba371e43dc3fb1c03ab1500e1f5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 22:35:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/06/565906/4/check/openstack-tox-pep8/fa4d751/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/06/565906/4/check/openstack-tox-py27/e51c0c8/ : FAILURE in 9m 53s\n- openstack-tox-py35 http://logs.openstack.org/06/565906/4/check/openstack-tox-py35/36842bd/ : FAILURE in 11m 11s\n- openstack-tox-py36 http://logs.openstack.org/06/565906/4/check/openstack-tox-py36/606a418/ : FAILURE in 9m 03s\n- openstack-tox-py37 http://logs.openstack.org/06/565906/4/check/openstack-tox-py37/c3061f1/ : FAILURE in 10m 27s\n- openstack-tox-docs http://logs.openstack.org/06/565906/4/check/openstack-tox-docs/3f45d0f/html/ : SUCCESS in 7m 58s\n- openstack-tox-cover http://logs.openstack.org/06/565906/4/check/openstack-tox-cover/f399522/ : FAILURE in 11m 46s (non-voting)\n- mistral-devstack http://logs.openstack.org/06/565906/4/check/mistral-devstack/5ea20bd/ : SUCCESS in 44m 22s\n- mistral-devstack-non-apache http://logs.openstack.org/06/565906/4/check/mistral-devstack-non-apache/7282bf2/ : SUCCESS in 47m 00s\n- mistral-devstack-kombu http://logs.openstack.org/06/565906/4/check/mistral-devstack-kombu/499c3ae/ : SUCCESS in 48m 02s\n- mistral-tox-unit-mysql http://logs.openstack.org/06/565906/4/check/mistral-tox-unit-mysql/bc05bc2/ : FAILURE in 28m 17s\n- mistral-tox-unit-postgresql http://logs.openstack.org/06/565906/4/check/mistral-tox-unit-postgresql/dc3851b/ : FAILURE in 27m 42s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/06/565906/4/check/tripleo-ci-centos-7-scenario003-standalone/4155dd6/ : SUCCESS in 1h 11m 34s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/06/565906/4/check/tripleo-ci-centos-7-containers-multinode/aba7265/ : SUCCESS in 2h 06m 07s\n- mistral-rally-task http://logs.openstack.org/06/565906/4/check/mistral-rally-task/8eb7290/ : SUCCESS in 1h 08m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/565906/4/check/openstack-tox-lower-constraints/bc31a9b/ : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":4},{"id":"691246c26b98a320f283138c2132b5597201acd7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-04-08 09:26:40.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"cd0a5762ef751593275154b446a3dc0447c0ac40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 11:38:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/06/565906/5/check/openstack-tox-pep8/b9d1568/ : SUCCESS in 6m 31s\n- openstack-tox-py27 http://logs.openstack.org/06/565906/5/check/openstack-tox-py27/daed65e/ : SUCCESS in 9m 18s\n- openstack-tox-py35 http://logs.openstack.org/06/565906/5/check/openstack-tox-py35/83f0b42/ : SUCCESS in 9m 54s\n- openstack-tox-py36 http://logs.openstack.org/06/565906/5/check/openstack-tox-py36/abbc9cf/ : FAILURE in 9m 34s\n- openstack-tox-py37 http://logs.openstack.org/06/565906/5/check/openstack-tox-py37/24e2b34/ : SUCCESS in 10m 16s\n- openstack-tox-docs http://logs.openstack.org/06/565906/5/check/openstack-tox-docs/8fdbfcd/html/ : SUCCESS in 6m 57s\n- openstack-tox-cover http://logs.openstack.org/06/565906/5/check/openstack-tox-cover/0b07c8e/ : FAILURE in 10m 35s (non-voting)\n- mistral-devstack http://logs.openstack.org/06/565906/5/check/mistral-devstack/08cf4cb/ : SUCCESS in 35m 31s\n- mistral-devstack-non-apache http://logs.openstack.org/06/565906/5/check/mistral-devstack-non-apache/666e73b/ : SUCCESS in 54m 35s\n- mistral-devstack-kombu http://logs.openstack.org/06/565906/5/check/mistral-devstack-kombu/4972562/ : SUCCESS in 44m 37s\n- mistral-tox-unit-mysql http://logs.openstack.org/06/565906/5/check/mistral-tox-unit-mysql/dbb61fe/ : SUCCESS in 28m 50s\n- mistral-tox-unit-postgresql http://logs.openstack.org/06/565906/5/check/mistral-tox-unit-postgresql/1ad4353/ : FAILURE in 28m 11s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/06/565906/5/check/tripleo-ci-centos-7-scenario003-standalone/522f8d1/ : SUCCESS in 1h 11m 15s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/06/565906/5/check/tripleo-ci-centos-7-containers-multinode/734eefb/ : SUCCESS in 2h 03m 33s\n- mistral-rally-task http://logs.openstack.org/06/565906/5/check/mistral-rally-task/b2b1402/ : FAILURE in 1h 03m 18s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/06/565906/5/check/openstack-tox-lower-constraints/94a6058/ : FAILURE in 11m 34s","accounts_in_message":[],"_revision_number":5},{"id":"b97bce1b704419c957e8cda3cdd3155e2571ec75","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:37:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"406cf7d2f0f24ef552c895d1251216810c1292f9","revisions":{"95cfec4b58823aa6f688e9355e52b5a974ef7d93":{"kind":"REWORK","_number":1,"created":"2018-05-02 22:22:43.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/06/565906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/565906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/565906/1"}}},"commit":{"parents":[{"commit":"1e577ae73925f7703a5d4e4452855208d0788679","subject":"Merge \"Release note for workflow environment optimizations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e577ae73925f7703a5d4e4452855208d0788679"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-05-02 22:22:29.000000000","tz":240},"committer":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-05-02 22:22:29.000000000","tz":240},"subject":"Timeout policy must fail task without retry","message":"Timeout policy must fail task without retry\n\nAdded new `force` parameter to complete task function. The default\nvalue is False. If force equals True, task doesn\u0027t execute a after\ncomplete policy.\n\nChange-Id: I5ae6a922e1fde60ab80ebd7ca81a5aff5ca2df9b\nCloses-bug: #1767352\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95cfec4b58823aa6f688e9355e52b5a974ef7d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95cfec4b58823aa6f688e9355e52b5a974ef7d93"}]},"branch":"refs/heads/master"},"2483145d5755db0f84a5d3e6c70f3ba58a9454d0":{"kind":"REWORK","_number":2,"created":"2018-05-03 02:44:25.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/06/565906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/565906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/565906/2"}}},"commit":{"parents":[{"commit":"1e577ae73925f7703a5d4e4452855208d0788679","subject":"Merge \"Release note for workflow environment optimizations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e577ae73925f7703a5d4e4452855208d0788679"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-05-02 22:22:29.000000000","tz":240},"committer":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-05-03 02:43:53.000000000","tz":240},"subject":"Timeout policy must fail task without retry","message":"Timeout policy must fail task without retry\n\nAdded new `force` parameter to complete task function. The default\nvalue is False. If force equals True, task doesn\u0027t execute a after\ncomplete policy.\n\nChange-Id: I5ae6a922e1fde60ab80ebd7ca81a5aff5ca2df9b\nCloses-bug: #1767352\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2483145d5755db0f84a5d3e6c70f3ba58a9454d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2483145d5755db0f84a5d3e6c70f3ba58a9454d0"}]},"branch":"refs/heads/master"},"f0a00b94289fe41dac94a8b89990ad71333d6ad7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2018-05-06 10:31:29.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/06/565906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/565906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/565906/3"}}},"commit":{"parents":[{"commit":"1e577ae73925f7703a5d4e4452855208d0788679","subject":"Merge \"Release note for workflow environment optimizations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e577ae73925f7703a5d4e4452855208d0788679"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-05-02 22:22:29.000000000","tz":240},"committer":{"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","date":"2018-05-06 10:31:18.000000000","tz":0},"subject":"WIP: Timeout policy must fail task without retry","message":"WIP: Timeout policy must fail task without retry\n\nAdded new `force` parameter to complete task function. The default\nvalue is False. If force equals True, task doesn\u0027t execute a after\ncomplete policy.\n\nChange-Id: I5ae6a922e1fde60ab80ebd7ca81a5aff5ca2df9b\nCloses-bug: #1767352\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0a00b94289fe41dac94a8b89990ad71333d6ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f0a00b94289fe41dac94a8b89990ad71333d6ad7"}]},"branch":"refs/heads/master"},"8b2e0d8dc412d48b220f764d20e05c98ca598b1b":{"kind":"REWORK","_number":4,"created":"2019-03-27 20:23:46.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/06/565906/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/565906/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/565906/4"}}},"commit":{"parents":[{"commit":"d332d7fefea8711d6c30cfcea80b8fef2767be2c","subject":"Add release notes for engine optimizations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d332d7fefea8711d6c30cfcea80b8fef2767be2c"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-05-02 22:22:29.000000000","tz":240},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-03-27 20:22:51.000000000","tz":180},"subject":"Timeout policy must fail task without retry","message":"Timeout policy must fail task without retry\n\nAdded new `force` parameter to complete task function. The default\nvalue is False. If force equals True, task doesn\u0027t execute a after\ncomplete policy.\n\nChange-Id: I5ae6a922e1fde60ab80ebd7ca81a5aff5ca2df9b\nCloses-bug: #1767352\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8b2e0d8dc412d48b220f764d20e05c98ca598b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8b2e0d8dc412d48b220f764d20e05c98ca598b1b"}]},"branch":"refs/heads/master"},"406cf7d2f0f24ef552c895d1251216810c1292f9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-04-08 09:26:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/06/565906/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/06/565906/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/06/565906/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/06/565906/5"}}},"commit":{"parents":[{"commit":"0775d2cc81415e03b787c1175fe790161dca5f08","subject":"Merge \"Add release note for I04ba85488b27cb05c3b81ad8c973c3cc3fe56d36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0775d2cc81415e03b787c1175fe790161dca5f08"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-05-02 22:22:29.000000000","tz":240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-04-08 09:26:40.000000000","tz":0},"subject":"Timeout policy must fail task without retry","message":"Timeout policy must fail task without retry\n\nAdded new `force` parameter to complete task function. The default\nvalue is False. If force equals True, task doesn\u0027t execute a after\ncomplete policy.\n\nChange-Id: I5ae6a922e1fde60ab80ebd7ca81a5aff5ca2df9b\nCloses-bug: #1767352\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/406cf7d2f0f24ef552c895d1251216810c1292f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/406cf7d2f0f24ef552c895d1251216810c1292f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
