)]}'
{"id":"openstack%2Fmistral~599288","triplet_id":"openstack%2Fmistral~master~Ib8bd3cf5c314ab4529260736cc0b7d792ce53171","project":"openstack/mistral","branch":"master","topic":"bug/1790441","hashtags":[],"change_id":"Ib8bd3cf5c314ab4529260736cc0b7d792ce53171","subject":"New rerun events","status":"MERGED","created":"2018-09-03 06:11:23.000000000","updated":"2019-07-18 11:23:50.000000000","submitted":"2019-07-18 11:23:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"599288-1563449030595-ef2d6f53","meta_rev_id":"d1391852596b11e30347d71b16870b9a30ee8294","_number":599288,"virtual_id_number":599288,"owner":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-07-18 11:23:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-07-17 10:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-07-18 09:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-07-18 09:31:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 09:21:38.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-06-13 08:18:04.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-06-13 08:18:04.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-06-13 08:18:04.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-07-17 10:57:40.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-07-18 09:31:34.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-07-18 11:23:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59bd1ef9e7ab5996411ddd69412189c52866803a","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-09-03 06:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59e8631c1d46e4dcb670684116023e5a3ea2de8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 08:51:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/599288/1/check/openstack-tox-pep8/4bfc523/ : SUCCESS in 6m 58s\n- openstack-tox-py27 http://logs.openstack.org/88/599288/1/check/openstack-tox-py27/207d89f/ : SUCCESS in 9m 55s\n- openstack-tox-py35 http://logs.openstack.org/88/599288/1/check/openstack-tox-py35/b599556/ : SUCCESS in 10m 40s\n- openstack-tox-py36 http://logs.openstack.org/88/599288/1/check/openstack-tox-py36/922d16d/ : SUCCESS in 9m 29s\n- openstack-tox-docs http://logs.openstack.org/88/599288/1/check/openstack-tox-docs/c0ef26e/html/ : SUCCESS in 5m 39s\n- openstack-tox-cover http://logs.openstack.org/88/599288/1/check/openstack-tox-cover/3f0995f/cover/ : SUCCESS in 11m 37s (non-voting)\n- mistral-devstack http://logs.openstack.org/88/599288/1/check/mistral-devstack/d28a86a/ : SUCCESS in 45m 24s\n- mistral-devstack-non-apache http://logs.openstack.org/88/599288/1/check/mistral-devstack-non-apache/cd57d90/ : SUCCESS in 40m 59s\n- mistral-devstack-kombu http://logs.openstack.org/88/599288/1/check/mistral-devstack-kombu/ddf4706/ : SUCCESS in 44m 40s\n- mistral-tox-unit-mysql http://logs.openstack.org/88/599288/1/check/mistral-tox-unit-mysql/a7ea4dc/ : SUCCESS in 31m 23s\n- mistral-tox-unit-postgresql http://logs.openstack.org/88/599288/1/check/mistral-tox-unit-postgresql/a2d5959/ : SUCCESS in 28m 38s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/88/599288/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/56d8f29/ : SUCCESS in 2h 36m 29s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/88/599288/1/check/tripleo-ci-centos-7-undercloud-containers/f4a8e54/ : SUCCESS in 2h 02m 46s\n- mistral-rally-task http://logs.openstack.org/88/599288/1/check/mistral-rally-task/aae00f5/ : SUCCESS in 1h 03m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/599288/1/check/openstack-tox-lower-constraints/2497f29/ : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d92aab34ed5bc0f7fd980141962f71065aa07a4b","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-06 22:32:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92a2498312de28c1807abc8c624f75f5a8b03c2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:48:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/88/599288/2/check/openstack-tox-pep8/02d7f54/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/88/599288/2/check/openstack-tox-py27/1ad18d1/ : SUCCESS in 8m 33s\n- openstack-tox-py36 http://logs.openstack.org/88/599288/2/check/openstack-tox-py36/bf91042/ : SUCCESS in 9m 19s\n- openstack-tox-py37 http://logs.openstack.org/88/599288/2/check/openstack-tox-py37/72622ce/ : SUCCESS in 9m 19s\n- openstack-tox-docs http://logs.openstack.org/88/599288/2/check/openstack-tox-docs/84371e3/html/ : SUCCESS in 6m 54s\n- openstack-tox-cover http://logs.openstack.org/88/599288/2/check/openstack-tox-cover/9d4c2f3/cover/ : SUCCESS in 8m 56s (non-voting)\n- mistral-devstack http://logs.openstack.org/88/599288/2/check/mistral-devstack/b379f03/ : SUCCESS in 48m 37s\n- mistral-devstack-non-apache http://logs.openstack.org/88/599288/2/check/mistral-devstack-non-apache/9e9da07/ : SUCCESS in 43m 45s\n- mistral-devstack-kombu http://logs.openstack.org/88/599288/2/check/mistral-devstack-kombu/c40b022/ : SUCCESS in 42m 22s\n- mistral-tox-unit-mysql http://logs.openstack.org/88/599288/2/check/mistral-tox-unit-mysql/5657484/ : FAILURE in 36m 38s\n- mistral-tox-unit-postgresql http://logs.openstack.org/88/599288/2/check/mistral-tox-unit-postgresql/1c7a5e4/ : FAILURE in 34m 50s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/88/599288/2/check/tripleo-ci-centos-7-scenario003-standalone/07c922c/ : SUCCESS in 1h 19m 42s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/88/599288/2/check/tripleo-ci-centos-7-containers-multinode/601f6cc/ : SUCCESS in 2h 05m 56s\n- mistral-rally-task http://logs.openstack.org/88/599288/2/check/mistral-rally-task/90ee8c2/ : SUCCESS in 1h 07m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/599288/2/check/openstack-tox-lower-constraints/8237bf0/ : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e45295b25110ac90a8dbc83ae90b68c05d2491d9","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-06-07 09:21:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"20641e4ea2339b8fc974d2471e7449957d267e06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:45:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/599288/2/check/openstack-tox-pep8/72c2df0/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/88/599288/2/check/openstack-tox-py27/e9d2ea9/ : SUCCESS in 9m 05s\n- openstack-tox-py36 http://logs.openstack.org/88/599288/2/check/openstack-tox-py36/04afaef/ : SUCCESS in 8m 46s\n- openstack-tox-py37 http://logs.openstack.org/88/599288/2/check/openstack-tox-py37/cec4315/ : SUCCESS in 9m 41s\n- openstack-tox-docs http://logs.openstack.org/88/599288/2/check/openstack-tox-docs/e4a7551/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/88/599288/2/check/openstack-tox-cover/aaa9bae/cover/ : SUCCESS in 10m 03s (non-voting)\n- mistral-devstack http://logs.openstack.org/88/599288/2/check/mistral-devstack/bca4bec/ : SUCCESS in 47m 40s\n- mistral-devstack-non-apache http://logs.openstack.org/88/599288/2/check/mistral-devstack-non-apache/c929239/ : SUCCESS in 46m 52s\n- mistral-devstack-kombu http://logs.openstack.org/88/599288/2/check/mistral-devstack-kombu/a1e8827/ : SUCCESS in 54m 16s\n- mistral-tox-unit-mysql http://logs.openstack.org/88/599288/2/check/mistral-tox-unit-mysql/7b95a54/ : SUCCESS in 27m 58s\n- mistral-tox-unit-postgresql http://logs.openstack.org/88/599288/2/check/mistral-tox-unit-postgresql/675e24f/ : SUCCESS in 26m 20s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/88/599288/2/check/tripleo-ci-centos-7-scenario003-standalone/56b3d39/ : SUCCESS in 1h 10m 39s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/88/599288/2/check/tripleo-ci-centos-7-containers-multinode/b3635ca/ : SUCCESS in 2h 03m 54s\n- mistral-rally-task http://logs.openstack.org/88/599288/2/check/mistral-rally-task/0a73148/ : SUCCESS in 1h 02m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/599288/2/check/openstack-tox-lower-constraints/0a7865e/ : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a5decf7cf0ec9caa6256b2acbc432e370607ea64","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-07-11 23:29:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bab6658b185a6c88c5b894ba0273186a8ce6cecc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 01:57:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/599288/3/check/openstack-tox-pep8/76f640d/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/88/599288/3/check/openstack-tox-py27/4cde08b/ : SUCCESS in 7m 57s\n- openstack-tox-py36 http://logs.openstack.org/88/599288/3/check/openstack-tox-py36/9896fbd/ : SUCCESS in 8m 21s\n- openstack-tox-py37 http://logs.openstack.org/88/599288/3/check/openstack-tox-py37/1db984a/ : SUCCESS in 8m 49s\n- openstack-tox-docs http://logs.openstack.org/88/599288/3/check/openstack-tox-docs/8a8fc79/html/ : SUCCESS in 5m 01s\n- openstack-tox-cover http://logs.openstack.org/88/599288/3/check/openstack-tox-cover/c8d2d75/cover/ : SUCCESS in 8m 40s (non-voting)\n- mistral-devstack http://logs.openstack.org/88/599288/3/check/mistral-devstack/8adc7b6/ : SUCCESS in 45m 22s\n- mistral-devstack-non-apache http://logs.openstack.org/88/599288/3/check/mistral-devstack-non-apache/544ef5d/ : SUCCESS in 43m 15s\n- mistral-devstack-kombu http://logs.openstack.org/88/599288/3/check/mistral-devstack-kombu/0a1da5b/ : SUCCESS in 43m 11s\n- mistral-tox-unit-mysql http://logs.openstack.org/88/599288/3/check/mistral-tox-unit-mysql/f0cf981/ : SUCCESS in 28m 36s\n- mistral-tox-unit-postgresql http://logs.openstack.org/88/599288/3/check/mistral-tox-unit-postgresql/0856a7f/ : SUCCESS in 26m 30s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/88/599288/3/check/tripleo-ci-centos-7-scenario003-standalone/b013c3f/ : SUCCESS in 1h 16m 14s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/88/599288/3/check/tripleo-ci-centos-7-containers-multinode/72deb49/ : SUCCESS in 2h 12m 53s\n- mistral-rally-task http://logs.openstack.org/88/599288/3/check/mistral-rally-task/216c2d1/ : SUCCESS in 59m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/88/599288/3/check/openstack-tox-lower-constraints/d9b8434/ : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ed3c7f06c869d853814671b39273fb1fca431397","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-07-17 10:57:40.000000000","message":"Patch Set 3: Code-Review+2\n\nDoes the documentation contain some section on the notifications? I couldn\u0027t find such.","accounts_in_message":[],"_revision_number":3},{"id":"1a2e06a41c43ece9294d3fdb8d41a4ec40370c16","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-07-18 09:31:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c6509a5068dd5ffe61ffaf46c80c5e74b8e40224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 09:31:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f0c95a3ec38e92bf086c9c832b7a0f636c563de5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 11:23:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"605f76ddb2cb881982e3839de6f9171e13696363","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 11:23:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/599288/3/gate/openstack-tox-pep8/b61f027/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/88/599288/3/gate/openstack-tox-py27/a4f073f/ : SUCCESS in 7m 51s\n- openstack-tox-py36 http://logs.openstack.org/88/599288/3/gate/openstack-tox-py36/bccd0df/ : SUCCESS in 7m 42s\n- openstack-tox-py37 http://logs.openstack.org/88/599288/3/gate/openstack-tox-py37/762d34b/ : SUCCESS in 9m 23s\n- openstack-tox-docs http://logs.openstack.org/88/599288/3/gate/openstack-tox-docs/c7753f3/html/ : SUCCESS in 4m 33s\n- mistral-devstack http://logs.openstack.org/88/599288/3/gate/mistral-devstack/b45f8e4/ : SUCCESS in 47m 45s\n- mistral-devstack-non-apache http://logs.openstack.org/88/599288/3/gate/mistral-devstack-non-apache/b0ac50a/ : SUCCESS in 44m 43s\n- mistral-tox-unit-mysql http://logs.openstack.org/88/599288/3/gate/mistral-tox-unit-mysql/fe580d0/ : SUCCESS in 27m 59s\n- mistral-tox-unit-postgresql http://logs.openstack.org/88/599288/3/gate/mistral-tox-unit-postgresql/dfa8ef1/ : SUCCESS in 26m 52s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/88/599288/3/gate/tripleo-ci-centos-7-scenario003-standalone/6a242c1/ : SUCCESS in 1h 24m 37s\n- mistral-devstack-kombu http://logs.openstack.org/88/599288/3/gate/mistral-devstack-kombu/e173134/ : SUCCESS in 59m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/599288/3/gate/openstack-tox-lower-constraints/e914ec7/ : SUCCESS in 9m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/88/599288/3/gate/tripleo-ci-centos-7-undercloud-containers/6490b48/ : SUCCESS in 1h 43m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aed90bab7e5989b172c6fad373c0a50628a05a8f","revisions":{"57674e639a95169672d01a6c7995041d18015c6d":{"kind":"REWORK","_number":1,"created":"2018-09-03 06:11:23.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/88/599288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/88/599288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/88/599288/1"}}},"commit":{"parents":[{"commit":"4a9ceaf47d24218e04c8c573fc144582e8553e73","subject":"switch documentation job to new PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a9ceaf47d24218e04c8c573fc144582e8553e73"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-03 06:10:44.000000000","tz":240},"committer":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-03 06:10:44.000000000","tz":240},"subject":"New rerun events","message":"New rerun events\n\nCurrently, all workflow actions, for example, start, pause, resume,\nstop have own events except rerun. Even when we rerun a task, event\n\"TASK_LAUNCHED\" doesn\u0027t propagate to a parent workflow and task.\n\nAdd new WORKFLOW_RERUN and TASK_RERUN events. These events propagate\nto a parent workflow and task. You can find a example of new events in\nthe following tests:\n* test_notify_rerun_task\n* test_notify_rerun_nested_workflow\n\nThe recursive parameter was removed in the workflow set_state method\nbecause it only is used in the rerun method. Recursive rerun was\nextracted in the separate function.\n\nChange-Id: Ib8bd3cf5c314ab4529260736cc0b7d792ce53171\nCloses-Bug: #1790441\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/57674e639a95169672d01a6c7995041d18015c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/57674e639a95169672d01a6c7995041d18015c6d"}]},"branch":"refs/heads/master"},"20170e1536786c535084a095035444848ba4a960":{"kind":"REWORK","_number":2,"created":"2019-06-06 22:32:32.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/88/599288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/88/599288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/88/599288/2"}}},"commit":{"parents":[{"commit":"f85caa02c9aaa64db4300b43854327d132a71ed4","subject":"Merge \"Skip context evaluation for non-conditional transitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f85caa02c9aaa64db4300b43854327d132a71ed4"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-03 06:10:44.000000000","tz":240},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-06-06 22:30:45.000000000","tz":180},"subject":"New rerun events","message":"New rerun events\n\nCurrently, all workflow actions, for example, start, pause, resume,\nstop have own events except rerun. Even when we rerun a task, event\n\"TASK_LAUNCHED\" doesn\u0027t propagate to a parent workflow and task.\n\nAdd new WORKFLOW_RERUN and TASK_RERUN events. These events propagate\nto a parent workflow and task. You can find a example of new events in\nthe following tests:\n* test_notify_rerun_task\n* test_notify_rerun_nested_workflow\n\nThe recursive parameter was removed in the workflow set_state method\nbecause it only is used in the rerun method. Recursive rerun was\nextracted in the separate function.\n\nChange-Id: Ib8bd3cf5c314ab4529260736cc0b7d792ce53171\nCloses-Bug: #1790441\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\nCo-Authored-By: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/20170e1536786c535084a095035444848ba4a960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/20170e1536786c535084a095035444848ba4a960"}]},"branch":"refs/heads/master"},"aed90bab7e5989b172c6fad373c0a50628a05a8f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-11 23:29:47.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/88/599288/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/88/599288/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/88/599288/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/88/599288/3"}}},"commit":{"parents":[{"commit":"6a9aa4c295b51e9b11ab4a30e5c002bcc2924609","subject":"Merge \"Initialize the lazily loaded execution \"input\" field in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6a9aa4c295b51e9b11ab4a30e5c002bcc2924609"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-03 06:10:44.000000000","tz":240},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2019-07-11 23:29:47.000000000","tz":0},"subject":"New rerun events","message":"New rerun events\n\nCurrently, all workflow actions, for example, start, pause, resume,\nstop have own events except rerun. Even when we rerun a task, event\n\"TASK_LAUNCHED\" doesn\u0027t propagate to a parent workflow and task.\n\nAdd new WORKFLOW_RERUN and TASK_RERUN events. These events propagate\nto a parent workflow and task. You can find a example of new events in\nthe following tests:\n* test_notify_rerun_task\n* test_notify_rerun_nested_workflow\n\nThe recursive parameter was removed in the workflow set_state method\nbecause it only is used in the rerun method. Recursive rerun was\nextracted in the separate function.\n\nChange-Id: Ib8bd3cf5c314ab4529260736cc0b7d792ce53171\nCloses-Bug: #1790441\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\nCo-Authored-By: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aed90bab7e5989b172c6fad373c0a50628a05a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aed90bab7e5989b172c6fad373c0a50628a05a8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
