)]}'
{"id":"openstack%2Fmistral~601501","triplet_id":"openstack%2Fmistral~master~I149b2e1dff7f86bd356f4dd2f758659469e6a4a8","project":"openstack/mistral","branch":"master","topic":"bug/1789384","hashtags":[],"change_id":"I149b2e1dff7f86bd356f4dd2f758659469e6a4a8","subject":"Docs improvements: task timeout, global context, Docker and jinja","status":"MERGED","created":"2018-09-11 06:35:11.000000000","updated":"2019-05-03 18:35:02.000000000","submitted":"2019-05-03 18:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"601501-1556908502241-13d10e42","meta_rev_id":"785e620415192324025bbe818ee864eef1016cc5","_number":601501,"virtual_id_number":601501,"owner":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-03 18:35:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-05-03 17:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-05-03 17:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-12 11:41:41.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2018-11-12 11:57:11.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2018-12-13 13:06:55.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2018-12-14 02:44:53.000000000","updated_by":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"reviewer":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"state":"REVIEWER"},{"updated":"2019-05-03 17:35:42.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-03 18:35:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f12490d4723f37fb1531bdc3cfe77745bc68f250","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-09-11 06:35:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57a129323f4d58f482b24817d0e11a69ae2e88de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-11 07:08:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/01/601501/1/check/openstack-tox-pep8/4fcdd2d/ : FAILURE in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/01/601501/1/check/openstack-tox-docs/c754c91/ : FAILURE in 4m 37s\n- openstack-tox-cover http://logs.openstack.org/01/601501/1/check/openstack-tox-cover/2bbe983/cover/ : SUCCESS in 9m 18s (non-voting)\n- mistral-tox-unit-mysql http://logs.openstack.org/01/601501/1/check/mistral-tox-unit-mysql/cfeeea2/ : SUCCESS in 31m 51s\n- mistral-tox-unit-postgresql http://logs.openstack.org/01/601501/1/check/mistral-tox-unit-postgresql/97acde6/ : SUCCESS in 27m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3091e5f5772eaa9722b432cf2b58803d5b3b1d58","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-09-11 08:14:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89356c086cad0e7166d4e8f98c47c80147aa2e91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-11 08:45:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/01/601501/2/check/openstack-tox-pep8/f2a0498/ : FAILURE in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/01/601501/2/check/openstack-tox-docs/36220f7/html/ : SUCCESS in 4m 55s\n- openstack-tox-cover http://logs.openstack.org/01/601501/2/check/openstack-tox-cover/05bc4e2/cover/ : SUCCESS in 9m 17s (non-voting)\n- mistral-tox-unit-mysql http://logs.openstack.org/01/601501/2/check/mistral-tox-unit-mysql/f353b9c/ : SUCCESS in 29m 37s\n- mistral-tox-unit-postgresql http://logs.openstack.org/01/601501/2/check/mistral-tox-unit-postgresql/e829180/ : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6d1a5c660798fa77aa6ee17b9b169000555137c6","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2018-09-12 11:41:41.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nLooks great! I\u0027ve not managed to read through it all yet, I\u0027ll do another pass today/tomorrow.","accounts_in_message":[],"_revision_number":2},{"id":"a1bae00e341d4a2db99e73152e23c8fecd26ebf4","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-10-15 22:03:18.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c98e91e2fbffaa33f3bb8afb648e8c6b2c8578b3","author":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"date":"2018-10-15 22:03:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d693d7ce8d34d564e835db7ad4cdd353338a8901","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-16 00:43:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/601501/3/check/openstack-tox-pep8/cb6b6b3/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/01/601501/3/check/openstack-tox-docs/faa43bb/html/ : SUCCESS in 5m 23s\n- openstack-tox-cover http://logs.openstack.org/01/601501/3/check/openstack-tox-cover/baaa73f/cover/ : SUCCESS in 10m 08s (non-voting)\n- mistral-tox-unit-mysql http://logs.openstack.org/01/601501/3/check/mistral-tox-unit-mysql/beb2151/ : SUCCESS in 29m 54s\n- mistral-tox-unit-postgresql http://logs.openstack.org/01/601501/3/check/mistral-tox-unit-postgresql/e019c20/ : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ac045f281a26e83d9abe26650b07bdd654e6c2ee","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2018-11-12 11:57:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"943f9072eabcfad3cdc99033127ce1dc93085000","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2018-12-13 13:06:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b22800fdd0fc0952a0f0d540213b3d16353282b9","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2018-12-14 02:44:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"86c0f357b38eb524d1af99ff8ce8e1dfdcbd606b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-03 17:25:03.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c444967b401bd031fb71b08ec4c6f1ec9570ba1f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-03 17:27:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2570640f93af96d3b4bfad0b6d098f251ed175a4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-05-03 17:35:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nGreat patch! Sorry for not reviewing it for so long. Since other people already gave their positive feedback on it I won\u0027t be waiting for another +2 and will approve it now.","accounts_in_message":[],"_revision_number":5},{"id":"efbdfd68b726e11f1826fc44eb33039f50afcab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:00:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/601501/5/check/openstack-tox-pep8/b00bd80/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/01/601501/5/check/openstack-tox-docs/f3ae0ee/html/ : SUCCESS in 5m 27s\n- openstack-tox-cover http://logs.openstack.org/01/601501/5/check/openstack-tox-cover/555a4f0/cover/ : SUCCESS in 8m 38s (non-voting)\n- mistral-tox-unit-mysql http://logs.openstack.org/01/601501/5/check/mistral-tox-unit-mysql/27d3483/ : SUCCESS in 28m 07s\n- mistral-tox-unit-postgresql http://logs.openstack.org/01/601501/5/check/mistral-tox-unit-postgresql/fd71e1a/ : SUCCESS in 27m 43s","accounts_in_message":[],"_revision_number":5},{"id":"86715318b09d755080100dbfc8e885c2a93ea504","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:00:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3b8ab7798f28715e8d0375fa5acdd2d1b0099893","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:35:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"380de5200ae01f81722244d813d3d064b3cf893e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:35:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/01/601501/5/gate/openstack-tox-pep8/79f2090/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/01/601501/5/gate/openstack-tox-docs/fc846ab/html/ : SUCCESS in 6m 03s\n- mistral-tox-unit-mysql http://logs.openstack.org/01/601501/5/gate/mistral-tox-unit-mysql/933448d/ : SUCCESS in 28m 14s\n- mistral-tox-unit-postgresql http://logs.openstack.org/01/601501/5/gate/mistral-tox-unit-postgresql/15b9b57/ : SUCCESS in 26m 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"09cd21d56106f417305b10c2f8f64568e916e0bd","revisions":{"615804cf98767a63bb1142c52190e70ddd1bd485":{"kind":"REWORK","_number":1,"created":"2018-09-11 06:35:11.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/01/601501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/01/601501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/01/601501/1"}}},"commit":{"parents":[{"commit":"604d905c24f5755b608d146a467bbfb833930e9b","subject":"Merge \"Add release note for auth_context bugfix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/604d905c24f5755b608d146a467bbfb833930e9b"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-11 06:33:59.000000000","tz":240},"committer":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-11 06:33:59.000000000","tz":240},"subject":"Docs improvements: task timeout, global context, Docker and jinja","message":"Docs improvements: task timeout, global context, Docker and jinja\n\nA not obvious point for users is the task does not retry after a\ntimeout is triggered. Added clarification in the retry section.\n\nThe documentation contains enough example with Jinja usage. Improved\nonly the create_vm workflow definition.\n\nAdded global publishing to the doc. It is brash copy-paste from\nhttps://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html without mention of atomic publish.\n\nMove Docker guides to the installation section.\n\nChange-Id: I149b2e1dff7f86bd356f4dd2f758659469e6a4a8\nCloses-Bug: #1789384\nCloses-Bug: #1690156\nCloses-Bug: #1779244\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/615804cf98767a63bb1142c52190e70ddd1bd485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/615804cf98767a63bb1142c52190e70ddd1bd485"}]},"branch":"refs/heads/master"},"1f1126d7d9e6c3bebe69aadd588b029513d6c69d":{"kind":"REWORK","_number":2,"created":"2018-09-11 08:14:41.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/01/601501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/01/601501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/01/601501/2"}}},"commit":{"parents":[{"commit":"604d905c24f5755b608d146a467bbfb833930e9b","subject":"Merge \"Add release note for auth_context bugfix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/604d905c24f5755b608d146a467bbfb833930e9b"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-11 06:33:59.000000000","tz":240},"committer":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-11 08:14:21.000000000","tz":240},"subject":"Docs improvements: task timeout, global context, Docker and jinja","message":"Docs improvements: task timeout, global context, Docker and jinja\n\nA not obvious point for users is the task does not retry after a\ntimeout is triggered. Added clarification in the retry section.\n\nThe documentation contains enough example with Jinja usage. Improved\nonly the create_vm workflow definition.\n\nAdded global publishing to the doc. It is brash copy-paste from\nhttps://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html without mention of atomic publish.\n\nMove Docker guides to the installation section.\n\nChange-Id: I149b2e1dff7f86bd356f4dd2f758659469e6a4a8\nCloses-Bug: #1789384\nCloses-Bug: #1690156\nCloses-Bug: #1779244\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f1126d7d9e6c3bebe69aadd588b029513d6c69d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1f1126d7d9e6c3bebe69aadd588b029513d6c69d"}]},"branch":"refs/heads/master"},"601d1fe0925f078751bee98c6bcfe49bbb780174":{"kind":"REWORK","_number":3,"created":"2018-10-15 22:03:43.000000000","uploader":{"_account_id":27008,"name":"Vitalii Solodilov","email":"Mcdkr@yandex.ru","username":"mcdoker18"},"ref":"refs/changes/01/601501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/01/601501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/01/601501/3"}}},"commit":{"parents":[{"commit":"604d905c24f5755b608d146a467bbfb833930e9b","subject":"Merge \"Add release note for auth_context bugfix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/604d905c24f5755b608d146a467bbfb833930e9b"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-11 06:33:59.000000000","tz":240},"committer":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-10-15 22:01:20.000000000","tz":240},"subject":"Docs improvements: task timeout, global context, Docker and jinja","message":"Docs improvements: task timeout, global context, Docker and jinja\n\nA not obvious point for users is the task does not retry after a\ntimeout is triggered. Added clarification in the retry section.\n\nThe documentation contains enough example with Jinja usage. Improved\nonly the create_vm workflow definition.\n\nAdded global publishing to the doc. It is brash copy-paste from\nhttps://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html without mention of atomic publish.\n\nMove Docker guides to the installation section.\n\nChange-Id: I149b2e1dff7f86bd356f4dd2f758659469e6a4a8\nCloses-Bug: #1789384\nCloses-Bug: #1690156\nCloses-Bug: #1779244\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/601d1fe0925f078751bee98c6bcfe49bbb780174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/601d1fe0925f078751bee98c6bcfe49bbb780174"}]},"branch":"refs/heads/master"},"6e9c7ee48923ea357386eaa2a926f3c1def756e5":{"kind":"REWORK","_number":4,"created":"2019-05-03 17:25:03.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/01/601501/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/01/601501/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/01/601501/4"}}},"commit":{"parents":[{"commit":"604d905c24f5755b608d146a467bbfb833930e9b","subject":"Merge \"Add release note for auth_context bugfix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/604d905c24f5755b608d146a467bbfb833930e9b"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-11 06:33:59.000000000","tz":240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-05-03 17:24:53.000000000","tz":0},"subject":"Docs improvements: task timeout, global context, Docker and jinja","message":"Docs improvements: task timeout, global context, Docker and jinja\n\nA not obvious point for users is the task does not retry after a\ntimeout is triggered. Added clarification in the retry section.\n\nThe documentation contains enough example with Jinja usage. Improved\nonly the create_vm workflow definition.\n\nAdded global publishing to the doc. It is brash copy-paste from\nhttps://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html without mention of atomic publish.\n\nMove Docker guides to the installation section.\n\nChange-Id: I149b2e1dff7f86bd356f4dd2f758659469e6a4a8\nCloses-Bug: #1789384\nCloses-Bug: #1690156\nCloses-Bug: #1779244\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9c7ee48923ea357386eaa2a926f3c1def756e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9c7ee48923ea357386eaa2a926f3c1def756e5"}]},"branch":"refs/heads/master"},"09cd21d56106f417305b10c2f8f64568e916e0bd":{"kind":"REWORK","_number":5,"created":"2019-05-03 17:27:00.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/01/601501/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/01/601501/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/01/601501/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/01/601501/5"}}},"commit":{"parents":[{"commit":"604d905c24f5755b608d146a467bbfb833930e9b","subject":"Merge \"Add release note for auth_context bugfix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/604d905c24f5755b608d146a467bbfb833930e9b"}]}],"author":{"name":"Vitalii Solodilov","email":"mcdkr@yandex.ru","date":"2018-09-11 06:33:59.000000000","tz":240},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-05-03 17:26:57.000000000","tz":0},"subject":"Docs improvements: task timeout, global context, Docker and jinja","message":"Docs improvements: task timeout, global context, Docker and jinja\n\nA not obvious point for users is the task does not retry after a\ntimeout is triggered. Added clarification in the retry section.\n\nThe documentation contains enough example with Jinja usage. Improved\nonly the create_vm workflow definition.\n\nAdded global publishing to the doc. It is brash copy-paste from\nhttps://specs.openstack.org/openstack/mistral-specs/specs/pike/approved/advanced_publishing.html without mention of atomic publish.\n\nMove Docker guides to the installation section.\n\nChange-Id: I149b2e1dff7f86bd356f4dd2f758659469e6a4a8\nCloses-Bug: #1789384\nCloses-Bug: #1690156\nCloses-Bug: #1779244\nSigned-off-by: Vitalii Solodilov \u003cmcdkr@yandex.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/09cd21d56106f417305b10c2f8f64568e916e0bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/09cd21d56106f417305b10c2f8f64568e916e0bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
